)]}'
{"id":"openstack%2Freleases~954467","triplet_id":"openstack%2Freleases~master~I2b9b685145299772a55fd6114c01b7e33c06501b","project":"openstack/releases","branch":"master","topic":"caracal-stable","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-07-10 09:47:04.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-07-10 09:23:08.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-10 09:47:04.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2025-07-10 09:47:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b9b685145299772a55fd6114c01b7e33c06501b","subject":"Release networking-baremetal for stable/2024.1","status":"MERGED","created":"2025-07-09 12:45:02.000000000","updated":"2025-07-10 09:47:04.000000000","submitted":"2025-07-10 09:47:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954467-caracal-stable","meta_rev_id":"eff93625c4e5afb3f820c7758e97b2faac660c66","_number":954467,"virtual_id_number":954467,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 09:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-07-10 09:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-07-10 09:23:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-07-10 09:23:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-09 12:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-09 12:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-09 13:20:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2025-07-09 13:20:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-07-10 09:08:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"bc0b9c56c1b85c2642eea01e287dd41a82775618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-09 12:45:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6229614e193b72d19d200390f17cb3d2169bed5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 12:55:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e74b74ea9f6c4f938331d39826e76991\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69790cbb0594138ae9f1bff79f51cf5 : SUCCESS in 5m 56s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/bf59a7b472f0442e807f739f8cacf9f8 : SUCCESS in 7m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5873db4688c94f779ac48fba0525dbb7 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"dafc67aeca358279fbbe76823d4bc46c71523b7e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 07:46:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76a97d209c7bce767c760624d6d5b56f834a6f06","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 07:47:34.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"7a6c328485621aa3974e22d4b256df2f3f0972c4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-07-10 09:08:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0556e4879c713912a155e2f1b82f3bae3d40b7ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-10 09:21:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc7fa0a38b7febcdebc46953485021f7b51cf41d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-10 09:23:08.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ad031fa43258ecc44801fff394f2cccf0828e6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:32:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e9a77885aa4e3aa7b78a620a2ddf86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce77322e9abe4624b5e421754d3a09e5 : SUCCESS in 9m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/35f941d959c54aa6b8f3a1d0a8843f62 : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/96dcc88b67664958a8c77dd01598d0b5 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e0148e343c13f642f3ec8b306f81a48416beee1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:32:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed21ce7fe6be9de2a62e4540c4100a3b8f5a437c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:47:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b535319a46543299b63631706e14779\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bd47dde7bb4ebc84ad0c93d999f909 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7606194fad4339ae1144841844cf01 : SUCCESS in 3m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/21719e722ec64f7e9c214b78b6e8a0c6 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"eff93625c4e5afb3f820c7758e97b2faac660c66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:47:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"046e5f1431d22b1397658194ea94a47fbf4d4a40","revisions":{"ea529c4b621ffac1e9f0b680508542fd8095b4f6":{"kind":"REWORK","_number":1,"created":"2025-07-09 12:45:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/954467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/954467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/954467/1"}}},"commit":{"parents":[{"commit":"ad8cc01b13d82219060f1cbe4ca408554435bd6b","subject":"Merge \"Release heat 24.1.0 for epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ad8cc01b13d82219060f1cbe4ca408554435bd6b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-09 12:44:57.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-09 12:44:57.000000000","tz":120},"subject":"Release networking-baremetal for stable/2024.1","message":"Release networking-baremetal for stable/2024.1\n\nAs a preparation for the transition of 2024.1 Caracal to Unmaintained\n[1], this release picks up new commits to networking-baremetal since\nthe last release from stable/2024.1.\n\nThis is being proposed as a convenience to make sure that merged stable\nchanges are released. If the team is good with this going out, please\nrespond with a +1 to let the release team know that it is OK to proceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 6.3.0..8dc6eda\n8dc6eda prevent break on communications failure\n01203f3 avoid attribute error on bad password or config\n62ca32e Update TOX_CONSTRAINTS_FILE for stable/2024.1\n632c03d Update .gitreview for stable/2024.1\n\nChange-Id: I2b9b685145299772a55fd6114c01b7e33c06501b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ea529c4b621ffac1e9f0b680508542fd8095b4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ea529c4b621ffac1e9f0b680508542fd8095b4f6"}]},"branch":"refs/heads/master"},"046e5f1431d22b1397658194ea94a47fbf4d4a40":{"kind":"REWORK","_number":2,"created":"2025-07-10 09:21:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/954467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/954467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/954467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/954467/2"}}},"commit":{"parents":[{"commit":"ad8cc01b13d82219060f1cbe4ca408554435bd6b","subject":"Merge \"Release heat 24.1.0 for epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ad8cc01b13d82219060f1cbe4ca408554435bd6b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-09 12:44:57.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 09:21:14.000000000","tz":0},"subject":"Release networking-baremetal for stable/2024.1","message":"Release networking-baremetal for stable/2024.1\n\nAs a preparation for the transition of 2024.1 Caracal to Unmaintained\n[1], this release picks up new commits to networking-baremetal since\nthe last release from stable/2024.1.\n\nThis is being proposed as a convenience to make sure that merged stable\nchanges are released. If the team is good with this going out, please\nrespond with a +1 to let the release team know that it is OK to proceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 6.3.0..8dc6eda\n8dc6eda prevent break on communications failure\n01203f3 avoid attribute error on bad password or config\n62ca32e Update TOX_CONSTRAINTS_FILE for stable/2024.1\n632c03d Update .gitreview for stable/2024.1\n\nChange-Id: I2b9b685145299772a55fd6114c01b7e33c06501b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/046e5f1431d22b1397658194ea94a47fbf4d4a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/046e5f1431d22b1397658194ea94a47fbf4d4a40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
