)]}'
{"id":"openstack%2Freleases~960055","triplet_id":"openstack%2Freleases~master~I792612643b2ab156c459a83e3af3adbbe4104f55","project":"openstack/releases","branch":"master","topic":"flamingo-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-11 12:28:01.000000000","reason":"Change was submitted"},"14394":{"account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"last_update":"2025-09-10 00:05:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14394\u003e replied on the change","reason_account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"}}},"hashtags":[],"change_id":"I792612643b2ab156c459a83e3af3adbbe4104f55","subject":"Release adjutant-ui RC1 for 2025.2 Flamingo","status":"MERGED","created":"2025-09-08 13:46:27.000000000","updated":"2025-09-11 12:28:01.000000000","submitted":"2025-09-11 12:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960055-flamingo-rc1-deadline","meta_rev_id":"b43011bfe1f74020b881add70748b03ecd9b2df3","_number":960055,"virtual_id_number":960055,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-11 12:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-09-11 12:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-09-10 08:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-09-10 00:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-09-11 12:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-09-10 00:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 13:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-08 14:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 14:22:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"state":"REVIEWER"},{"updated":"2025-09-11 12:03:30.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c2d7e9e760799de8e2589c81ec20f0bcc4ac1d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-08 13:46:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cef3d5bf9a60dcf7a66553a4dbae28836942688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 14:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21f02a2091e04f6d9342472674b5c610\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0bbda119ae424797b398a4856ec57f : SUCCESS in 5m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/615116ccdf364cb8a6f01d3ec1ab57ef : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/39203d7b32d04bad8b390cb3c6a0da35 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"92e5c0f5a2f9744c6909eb84a197f60ac104975a","author":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"date":"2025-09-10 00:05:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cbcb46933fdee32b3070f00a121807eb9d9f6d38","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 00:05:38.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"64f025ad4c6c51a5d259abbc6f50f23c5ba9195a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 08:51:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b90f20d14391e4ba4bd237cdb6666625a0ff6932","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-09-11 12:03:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73950238801642d029f2963a08494de0d31a237c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:12:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23d2202533e4a78959c7f22ba0cec65dea6c46ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:28:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8391ff6ca12c465d818a7434d3d5017d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145f44534c854056be7e3f47acecd1d4 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f4305f6a794eb8b15164996ecebcb6 : SUCCESS in 5m 42s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/878626245b414d7da7b33a5b1264b22d : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b43011bfe1f74020b881add70748b03ecd9b2df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:28:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec0efa7612f8df382888e31e1e2888418643875b","revisions":{"ec0efa7612f8df382888e31e1e2888418643875b":{"kind":"REWORK","_number":1,"created":"2025-09-08 13:46:27.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/960055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/960055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/960055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/960055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/960055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/960055/1"}}},"commit":{"parents":[{"commit":"0f83de6df74f24ec20ff9dab29e345f24db3c886","subject":"Merge \"[oslo] Create 2025.2 branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f83de6df74f24ec20ff9dab29e345f24db3c886"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-08 13:46:22.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-08 13:46:22.000000000","tz":120},"subject":"Release adjutant-ui RC1 for 2025.2 Flamingo","message":"Release adjutant-ui RC1 for 2025.2 Flamingo\n\nThis proposed a RC1 release for adjutant-ui to get any merged changes\nout prior to the RC1 deadline for 2025.2 Flamingo going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready. If there is nothing\nto wait for, then please do respond on this patch with +1 as soon as\npossible.\n\nIn any case, if the patch gets no response, then this will be merged\non Friday, September 12th.\n\nChange-Id: I792612643b2ab156c459a83e3af3adbbe4104f55\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec0efa7612f8df382888e31e1e2888418643875b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec0efa7612f8df382888e31e1e2888418643875b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
