)]}'
{"id":"openstack%2Freleases~960083","triplet_id":"openstack%2Freleases~master~I3d5c8838f054f81f529ef2a1e57dd1a7bcae5ec8","project":"openstack/releases","branch":"master","topic":"flamingo-rc1-deadline","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-12 09:01:36.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-12 08:48:53.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"14394":{"account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"last_update":"2025-09-11 04:02:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14394\u003e replied on the change","reason_account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"}}},"hashtags":[],"change_id":"I3d5c8838f054f81f529ef2a1e57dd1a7bcae5ec8","subject":"Release magnum-ui RC1 for 2025.2 Flamingo","status":"MERGED","created":"2025-09-08 14:06:04.000000000","updated":"2025-09-12 12:02:24.000000000","submitted":"2025-09-12 09:01:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960083-flamingo-rc1-deadline","meta_rev_id":"fe7df3347bdfe9b0a591bcf923be5f7fe838bbe6","_number":960083,"virtual_id_number":960083,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 09:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-09-12 08:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-09-12 08:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-09-12 00:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-12 12:02:24.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-09-12 08:48:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-09-12 00:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 14:06:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-08 14:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 14:21:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-08 14:21:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"state":"REVIEWER"},{"updated":"2025-09-12 08:32:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-09-12 12:02:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-09-12 12:02:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"9da725eb85ecc1eb7cf5fd4e7402c9dfd468fe98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-08 14:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a329d9b1d507e6d4d663ae038eedce10390673c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 14:12:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0638e48646e843d48ba5c48cb8c06a47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fede72ffe3149f59475a6dab1e1831e : SUCCESS in 5m 48s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/eafb6b76e0b3470b8f704b82a47279ee : SUCCESS in 3m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f81c848dae741599b27c491c1c5dac6 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fd268f05506d052afa0b645d7300d99da853221c","author":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"date":"2025-09-11 04:02:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9b29cb322ab6b1ffe4a9c2120b74e89e1d76c8c0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 04:03:33.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"56d7663237e076bf78c6d75d15a8562ad9b509e7","author":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"date":"2025-09-11 04:04:02.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"467daf197ec39f012f18ae093992fc832eb045ef","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 04:04:38.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"a0436e294e4e9e68319ebee7211d0add694f269c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-11 08:15:30.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c987ecb416e90db43c7c3619bb4481762e46617a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 08:25:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea218379c8c4e3f8adba158f9c4c566\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa5c10146054330896b8efeef69fa9a : SUCCESS in 8m 06s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b64eb94a52844e4ba8e8eab77d334039 : SUCCESS in 3m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/863441fab9184b73a3e76bfaba592034 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"19ed59a78efd0b4c26e505034eab4f8d04847e8e","author":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"date":"2025-09-12 00:56:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c04ce9a04dfb589c8138a07e5efaed1690d0cb4e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 00:57:02.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a0174911caf6041e20b8eaf5d8601a3ad782d130","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-09-12 08:32:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"761e850eb9c5532d248600c09334f1f822c12ffd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-12 08:48:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c20e74603e3eadb929bbadcba1341b34ce914cbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:49:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"842fb866dbb3e2e3f7405dc4b9a6a794c2cbd080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 09:01:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75e85f0ace094c6089f2a7cd06472c3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02eedf2c31a64a6188811748e965e088 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce407af3e8240ffb5e41528aa1af5ee : SUCCESS in 3m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/0d1eb03044864f96b65c4e3bfc7da2e8 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cbca33890da79a4bf2fc84f8ab71b519a519f626","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 09:01:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a61bed1e34c9ef8ed63487eba7a6b8f3b71fe566","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 12:02:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe7df3347bdfe9b0a591bcf923be5f7fe838bbe6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-12 12:02:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1edfea55ff99e14f654e1e3ef71c06c8e6ddabd","revisions":{"d9c003260e5d35332b736b92ff3d5d2591b3263e":{"kind":"REWORK","_number":1,"created":"2025-09-08 14:06:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/83/960083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/960083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/960083/1"}}},"commit":{"parents":[{"commit":"0f83de6df74f24ec20ff9dab29e345f24db3c886","subject":"Merge \"[oslo] Create 2025.2 branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f83de6df74f24ec20ff9dab29e345f24db3c886"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-08 14:05:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-08 14:05:59.000000000","tz":120},"subject":"Release magnum-ui RC1 for 2025.2 Flamingo","message":"Release magnum-ui RC1 for 2025.2 Flamingo\n\nThis proposed a RC1 release for magnum-ui to get any merged changes\nout prior to the RC1 deadline for 2025.2 Flamingo going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready. If there is nothing\nto wait for, then please do respond on this patch with +1 as soon as\npossible.\n\nIn any case, if the patch gets no response, then this will be merged\non Friday, September 12th.\n\nChange-Id: I3d5c8838f054f81f529ef2a1e57dd1a7bcae5ec8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9c003260e5d35332b736b92ff3d5d2591b3263e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9c003260e5d35332b736b92ff3d5d2591b3263e"}]},"branch":"refs/heads/master"},"c1edfea55ff99e14f654e1e3ef71c06c8e6ddabd":{"kind":"REWORK","_number":2,"created":"2025-09-11 08:15:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/83/960083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/960083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/960083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/960083/2"}}},"commit":{"parents":[{"commit":"0f83de6df74f24ec20ff9dab29e345f24db3c886","subject":"Merge \"[oslo] Create 2025.2 branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f83de6df74f24ec20ff9dab29e345f24db3c886"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-08 14:05:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-11 08:15:29.000000000","tz":0},"subject":"Release magnum-ui RC1 for 2025.2 Flamingo","message":"Release magnum-ui RC1 for 2025.2 Flamingo\n\nThis proposed a RC1 release for magnum-ui to get any merged changes\nout prior to the RC1 deadline for 2025.2 Flamingo going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready. If there is nothing\nto wait for, then please do respond on this patch with +1 as soon as\npossible.\n\nIn any case, if the patch gets no response, then this will be merged\non Friday, September 12th.\n\nChange-Id: I3d5c8838f054f81f529ef2a1e57dd1a7bcae5ec8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1edfea55ff99e14f654e1e3ef71c06c8e6ddabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1edfea55ff99e14f654e1e3ef71c06c8e6ddabd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
