)]}'
{"id":"openstack%2Freleases~971276","triplet_id":"openstack%2Freleases~master~I3279f46878ab9fe17b212e16a9df335f7f36d221","project":"openstack/releases","branch":"master","topic":"release_osa","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-12-18 13:21:37.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-12-19 14:39:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3279f46878ab9fe17b212e16a9df335f7f36d221","subject":"[openstack-ansible] Final release for Caracal EOM","status":"ABANDONED","created":"2025-12-17 08:43:38.000000000","updated":"2025-12-19 14:39:04.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76ac5373e769ae00ff61ae34363260103cc514a4","_number":971276,"virtual_id_number":971276,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-17 08:58:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-12-17 08:44:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 08:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-17 10:10:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"756134f97dd95dafe57efca015e3f03bdb300cca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 08:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"974d668fb942277c96d0baa2f391972538b9c139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 08:44:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8ee72bc34345647b848b47747dd376b58b3cede","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 08:44:50.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"20054cc1d0db3a0ddd7e9b43d5789cd695ec6965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 08:58:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b36026774cca4c6b9c2d6aed161dadf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ad3d3f8bec4dff9cffc0384ca55d05 : SUCCESS in 8m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9d1df5f8ea694295b8cc372b6f759153 : FAILURE in 4m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7eba52875f60460f9bd061f1e9a9d36d : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cb48063dc4e28b94e394e2af29a7b669c89b4d29","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-17 10:10:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aaf17913ddeb381f6c2b0bbf62ef5307baf05d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 10:40:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13fabb9d5b284652883ba86efcf6d238\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f0f7ed09bc4e16858e88f0206bb3b0 : SUCCESS in 6m 55s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/145f2b21a94546d6a7062a6d6d85af24 : FAILURE in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0782304362e6411a8b24496c9d5e13fd : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3df1a4d1b957adfc7c5bbb438ff3499900449c86","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 13:23:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"381986d3183f130228f76cab3f66dd4b6ef4d746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 13:29:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c230046b136a43eba029431368c4d3c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ebf214e57642e591c9c0f1bfc23a63 : SUCCESS in 5m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1a0d43677d514a9b91b993622c81cfb3 : FAILURE in 4m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d429b6f9333a4d94964226db26a3eb68 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"4caf17b24e852d64351822576f98cd4c9547e9b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-17 15:05:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27fc7fe88711c07b6a0be6b9481fa5ff7fa93099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:17:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21ee7708f9d04591a7029cf2feb3bb7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb464c9e98ac4b82a28c2b8dd9b5c1db : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b76f12d5418848f982b801534d1e3b89 : FAILURE in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2feeb8924d6648dfbd5821316c8b5fed : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"038a2d11b7ed44d9fa503bc539dcb10675eec6b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-18 10:02:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c545ae5b41c81711e583a972b2abe7af46cdb59d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 10:15:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ca2a8d5b46475fac0a3f5be38231f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a93fe8b9944957b178cf660efb9bdd : SUCCESS in 6m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e0e22a52968d4635bf9afacf306f3282 : FAILURE in 3m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/939df49cd0f840babb129e76670f2db7 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"67329c7c9862d971c83cdbba711dc01ae71b97c6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-18 10:51:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d35f17ff89dc79983bcf1bbce483a85347fa70f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-18 13:12:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"808fa5b4e6c7a9b8d396d3fc2aa3a4589d9658ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-18 13:12:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81e9a93666ac99d553216ded19ca0e3ccd5177af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-18 13:21:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76ac5373e769ae00ff61ae34363260103cc514a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-19 14:39:04.000000000","message":"Abandoned\n\nabandoning as the child patch, that moves OSA\u0027s stable/2024.1 to unmaintained, has merged instead, which makes this impossible to merge.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f00a06f92a3dd0b0ef73dc3155e421654fdf339","revisions":{"c0538f576281bccd3c138c489097be1118d6a676":{"kind":"REWORK","_number":1,"created":"2025-12-17 08:43:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/971276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/971276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/971276/1"}}},"commit":{"parents":[{"commit":"cb2385b56f10c7b557c4620f81400c063bb22fc8","subject":"Merge \"Create new automaton release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb2385b56f10c7b557c4620f81400c063bb22fc8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 08:43:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 08:43:27.000000000","tz":60},"subject":"[openstack-ansible] Final release for Caracal EOM","message":"[openstack-ansible] Final release for Caracal EOM\n\nChange-Id: I3279f46878ab9fe17b212e16a9df335f7f36d221\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0538f576281bccd3c138c489097be1118d6a676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0538f576281bccd3c138c489097be1118d6a676"}]},"branch":"refs/heads/master"},"6f00a06f92a3dd0b0ef73dc3155e421654fdf339":{"kind":"REWORK","_number":2,"created":"2025-12-17 08:44:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/76/971276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/971276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/971276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/971276/2"}}},"commit":{"parents":[{"commit":"cb2385b56f10c7b557c4620f81400c063bb22fc8","subject":"Merge \"Create new automaton release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb2385b56f10c7b557c4620f81400c063bb22fc8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 08:43:27.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-12-17 08:43:56.000000000","tz":60},"subject":"[openstack-ansible] Final release for Caracal EOM","message":"[openstack-ansible] Final release for Caracal EOM\n\nChange-Id: I3279f46878ab9fe17b212e16a9df335f7f36d221\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f00a06f92a3dd0b0ef73dc3155e421654fdf339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f00a06f92a3dd0b0ef73dc3155e421654fdf339"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
