)]}'
{"id":"openstack%2Freleases~972214","triplet_id":"openstack%2Freleases~master~I0e9981f77304319a4015f954bb461c6a4aca69d7","project":"openstack/releases","branch":"master","topic":"ovsdbapp_2.15.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-08 13:19:31.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-01-07 09:06:57.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-08 13:19:31.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-08 13:19:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e9981f77304319a4015f954bb461c6a4aca69d7","subject":"Release ovsdbapp 2.15.0","status":"MERGED","created":"2026-01-05 16:15:37.000000000","updated":"2026-01-08 13:19:31.000000000","submitted":"2026-01-08 13:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972214-ovsdbapp_2.15.0","meta_rev_id":"f84bc26a7b92375aaefeb99b9537fbc1939b30fa","_number":972214,"virtual_id_number":972214,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-08 13:19:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-07 07:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-08 13:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2026-01-07 09:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-06 20:33:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-08 13:04:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-01-06 20:33:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 16:15:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-01-05 16:15:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-05 16:15:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-01-05 16:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-05 16:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-05 16:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-05 17:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 10:45:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-01-08 13:04:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"1a4e3dab19b2d07b55f701e2122add5cff124855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-05 16:15:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"443c557d2c7d0faf643697f7efa1f56048a39f87","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-05 17:10:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"356d9c70a99abf64c00f9397381dc1727865ba33","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:11:01.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"dd635c445e1362abfced551ab65ab178886e4aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56c0cb3e43d429381e0decb4fb2cefe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96e338958614e029d4e82b21b1a6b08 : SUCCESS in 6m 38s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/dd0f01501fed4c3dae3489e5c53dc8af : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cd248d81c01748eeb8845fad931cf782 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"750a735fa06cb672249c8496be84f9e9a66ff4f2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-06 10:45:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b597835a563084062c540667c0fd118dec340dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-06 20:32:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"36a17c5acc3d9ca8c276bbd97fee1699211db220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-06 20:33:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bb0897b5daf79019108614890941b9076ceb61fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-06 20:33:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f86226fd1f12b7c95017a8394b5613d5fcd6c94","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 20:33:44.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a9220199b789d3fa5a88ea4c271f654609e5d991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 20:44:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246612e1769c46ea9e1030778ef9f4be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9983c15c6d54b658b09674a4e660367 : SUCCESS in 5m 34s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1494b31dc1604e948fc873affac167ea : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5ed3651b897471d922582df1d8eb772 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"55025a871647c3137f359dd32e8975a1735b5f4b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 07:55:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a8ba552b62cd636c61e1fadfa938db7687dbf7b7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 07:57:51.000000000","message":"Topic ovsdbapp_2.14.2 removed","accounts_in_message":[],"_revision_number":3},{"id":"ead3969d73c496fd28c3736271665c5862dccd7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 07:57:58.000000000","message":"Topic set to ovsdbapp_2.15.0","accounts_in_message":[],"_revision_number":3},{"id":"7cbce1470715219b832853481088c335cc1891cf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-07 09:06:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ada1b570e3627a5de7862d35790c9005a8fd5377","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-01-08 13:04:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b13098748e516be0f2d216115f6652d14335cf89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:04:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"306cacb04ce0d75270d5cfca17c111665ecda822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:19:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54327bba1534a26a9c3d42ce5d548bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15b04c851254e61a1d8583b6a938f7e : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2043de10dd8949caa60a10f5e6734623 : SUCCESS in 7m 24s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a741571b3e144740a94a1d8e4e1975bd : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"f84bc26a7b92375aaefeb99b9537fbc1939b30fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:19:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fc3c3f53e5aed33dab019d3079dddf9c2070d3f7","revisions":{"e74fd262857644656788abe0d1f8adc0626a5b6a":{"kind":"REWORK","_number":1,"created":"2026-01-05 16:15:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/14/972214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/972214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/972214/1"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-05 16:14:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-05 16:14:57.000000000","tz":60},"subject":"Release ovsdbapp 2.14.2","message":"Release ovsdbapp 2.14.2\n\n$ git log --oneline --no-merges 2.14.1..b4b4645\nb4b4645 windows: Add close() method to WaitQueue\nce46cc4 Close the pipe when Connection instance is deleted\n84813d9 Add project badge\nd27a01b Migrate setup configuration to pyproject.toml\n13091b6 Drop workaround for Python \u003c 2.7.4\nd5600ac Update supported python versions\n65a4c17 add pyproject.toml to support pip 23.1\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0e9981f77304319a4015f954bb461c6a4aca69d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e74fd262857644656788abe0d1f8adc0626a5b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e74fd262857644656788abe0d1f8adc0626a5b6a"}]},"branch":"refs/heads/master"},"877bb1b01a357d797fae393eb945ce73e36b7fb1":{"kind":"REWORK","_number":2,"created":"2026-01-06 20:32:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/14/972214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/972214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/972214/2"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-05 16:14:57.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-06 20:32:27.000000000","tz":0},"subject":"Release ovsdbapp 2.14.2","message":"Release ovsdbapp 2.14.2\n\n$ git log --oneline --no-merges 2.14.1..b4b4645\nb4b4645 windows: Add close() method to WaitQueue\nce46cc4 Close the pipe when Connection instance is deleted\n84813d9 Add project badge\nd27a01b Migrate setup configuration to pyproject.toml\n13091b6 Drop workaround for Python \u003c 2.7.4\nd5600ac Update supported python versions\n65a4c17 add pyproject.toml to support pip 23.1\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0e9981f77304319a4015f954bb461c6a4aca69d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/877bb1b01a357d797fae393eb945ce73e36b7fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/877bb1b01a357d797fae393eb945ce73e36b7fb1"}]},"branch":"refs/heads/master"},"fc3c3f53e5aed33dab019d3079dddf9c2070d3f7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-06 20:33:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/14/972214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/972214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/972214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/972214/3"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-05 16:14:57.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-06 20:33:04.000000000","tz":0},"subject":"Release ovsdbapp 2.15.0","message":"Release ovsdbapp 2.15.0\n\n$ git log --oneline --no-merges 2.14.1..b4b4645\nb4b4645 windows: Add close() method to WaitQueue\nce46cc4 Close the pipe when Connection instance is deleted\n84813d9 Add project badge\nd27a01b Migrate setup configuration to pyproject.toml\n13091b6 Drop workaround for Python \u003c 2.7.4\nd5600ac Update supported python versions\n65a4c17 add pyproject.toml to support pip 23.1\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0e9981f77304319a4015f954bb461c6a4aca69d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc3c3f53e5aed33dab019d3079dddf9c2070d3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc3c3f53e5aed33dab019d3079dddf9c2070d3f7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
