)]}'
{"id":"openstack%2Freleases~972457","triplet_id":"openstack%2Freleases~master~I56e98f7405fb2467173e66c8867bc87f471fe45f","project":"openstack/releases","branch":"master","topic":"neutron_28.0.0.0b2","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-13 14:15:05.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2026-01-13 14:15:05.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-13 14:15:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56e98f7405fb2467173e66c8867bc87f471fe45f","subject":"Release Neutron 28.0.0.0b2","status":"MERGED","created":"2026-01-07 11:49:25.000000000","updated":"2026-01-13 14:15:05.000000000","submitted":"2026-01-13 14:15:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972457-neutron_28.0.0.0b2","meta_rev_id":"46d3525d263e9a846e561edc44077567abefeaa7","_number":972457,"virtual_id_number":972457,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-13 14:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-01-13 14:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-01-10 11:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-01-12 15:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-13 14:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-01-08 17:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 11:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-07 11:59:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-01-07 12:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 12:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-01-07 12:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-01-10 11:38:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-01-12 15:09:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-13 14:06:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c5856b596b5787c12b54bed4db45c50b34524519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 11:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98e061db26b565a946a01bab149044662ccc6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12520ea494d149ef8fab84cb5c68dd69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6c87126a6e4fbcace5b3c4c6c1f696 : SUCCESS in 6m 03s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9119346eb5ff4f6e88a02eac51a3e518 : SUCCESS in 2m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b1783d0eab1f4c3398d8362a45ec2488 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7397b3d4195ab25634a346f39b1e55e268f1c433","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 13:41:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1667bc53419d39e66c46fe26d7a7e8b66f9d7863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 15:52:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5986339b1a911d7ee31d8b426d1028145edfc37","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-07 15:52:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e61c007ffe7e17989ed7cacdb0d0edb26d67e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 16:00:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc978cc9191424ca4401649890f9d7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ea418c3ee04286b4c57300cd1b272a : SUCCESS in 3m 38s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e909dc918a614643a4cdda94b733079c : SUCCESS in 2m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/24da9e11e18c447b9d9d8b9534c33164 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"03f08997c7ddc518866d0a903f176b73ccc59ed1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-08 17:16:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f322352184afffa5595d9c22b113501e67f4cf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-08 17:16:41.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f2ce80fc26ad490847fc1c7cdd1c0cf521e2fa0b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:17:33.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"46fd33b6929bcf7e5563883e546f4d39551941dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:46:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be28af65bed44770baafa03b64ae3a68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097b56d354844f7e9ff93b0bb59e0656 : SUCCESS in 6m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/45e4bb0ce07244cab2e9650ed0f79490 : SUCCESS in 4m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ac3087871e8a4fde9ccdfc2ebbd470ae : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"7f84905c42c97333ba36e9e47ce83f4c50338ce5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-10 11:38:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79696ddd0964899382e9cffe4228a8487adb2896","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-12 15:09:10.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"45f7e69a8563828d12d7252ace4b82268694d092","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-13 14:06:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e21ecb7dd0facc74af8a373f594704f7332d59d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:06:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"afee70d3f141b596d54c1ab0a9661cd0d1015e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:15:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96528aafae324f50a9f6d78311ae3818\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71318d3f377943a09987c44f630250cb : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9839cf018bd6442588156cc573059add : SUCCESS in 7m 20s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7ec07358d0824a2da2f7fa227087168b : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"46d3525d263e9a846e561edc44077567abefeaa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 14:15:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d3983034f4cba04b75d8133a298d933b45c495e2","revisions":{"5414dbb0a3fa9c2183a0ff77475196a1f1c3f861":{"kind":"REWORK","_number":1,"created":"2026-01-07 11:49:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/972457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/972457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/972457/1"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:48:04.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:48:04.000000000","tz":60},"subject":"Release Neutron 28.0.0.0b2","message":"Release Neutron 28.0.0.0b2\n\nThis version contains the unit and functional changes that replace\n\"tenant_id\" by \"project_id\".\n\n$ git log --oneline --no-merges 28.0.0.0b1..af2aefeb87\ne1f95eea4a Fix duplicated checks to enforce usage of assertIs(Not)None\n487552c4de [Docs] Remove note about FWaaS not being supported by OVN backend\n185b5ffe2b Doc: add warnig for fullstack\nef78d8e978 bgp: Improve existing server Commands\n02b7aac66b Change test code to use project_id in requests\n29aee58292 Change test code to use project_id to create objects\n8bf2b5d318 Change test code to use project_id in dictionaries\nbc096adac1 Change some functional test code to use project_id\n32762f848e Move api-paste.ini, rootwrap.conf files\n680b697247 Change some unit test code to use project_id\n4808ad7901 Remove the \"nftables\" explicit CI jobs\n2faba4ede3 Refactor ovn-db-sync script\nc8d06727c3 bgp: Remove chassis indexing\n4d196d1cd1 Check if network is external in auto-allocate code\n9edf82b24d [OVN] Fix the `_gen_router_port_ext_ids` call\n0e9f221de1 Drop redundant target-version option\n995785b0e8 Remove redundant mock statement\n137949ee07 Search for active port binding when updating LSP host info\nc6b2c1d612 bgp: Introduce Main router and chassis router commands\nd251834cc6 Use NumaAffinityPoliciesEnumField from neutron-lib\n7fe015c871 bgp: Introduce service plugin\nd8ec541122 ruff: Enable E5 check\n28a1483b70 [Docs] Add doc about Virtual IP addresses in Neutron ML2/OVN\na991dd1d6b Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing\n634b457ccf Fix ignored metadata options\n084ffb5866 Remove a missing Windows configuration file\n099e90b2e4 [OVN] OVN agent configuration options list missing \"extensions\"\n75c8ff55ad [OVN] Add documentation/reno for the OVN external port N/S routing\ncdf75bdea7 iPXE over IPv6 supported since OVN v23.06.0\n9cd84180ba [FT] Add extra logs to ``test_create_bridges`` test\n31b78bf23e [FT] Wait for the manager to be created (2)\n64d35a7c12 Change some common test code to use project_id\n6f4917d2d3 [OVN] DNAT registers always have the FIP external MAC\n71ae26e352 Don\u0027t prevent setting CIDRs as allowed_address_pair for port\na924eb9faa [OVN] Add \u0027provider:physical_network\u0027 field if there is physnet\n26513762c1 Allow plugins to add periodics to maintenance worker\n45b8dfe6a8 [OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\n3df3d8ebd2 [eventlet-removal] Deprecate wsgi options\nf2d6eaa61c Bump ovsdbapp to version 2.14.1\nbdcee6d5e0 [docs] Add OVS/OVN min requirement table\n807e6a5cf6 Fix incorrect log message in ovn_client\n27bd4f687b Add `neutron-functional-with-ovsdbapp-master` job\ncdac238f9a Centralize oslo.service patching\n3199846d4c Increase `neutron-functional` job timeout to 2.5 hours\ncf41465164 Use common constant for neutron-ovn-agent process name\n7d908bf182 Add ovn-agent to metadata caching doc\n20585d865d Revert \"[eventlet-removal] Handle stop DHCP agent\"\nd95e3e5eaf Replace response body in after hook for 404 error\n5e92bbfa50 [OVN] Initialize OVN agent in ``start`` method\n5e626f8bda [OVN] Sync the LRP Gateway_Chassis with the network HCG\ne94e7447aa [OVN] The external networks GW chassis must the same as the GW LRP\n78593ce9b8 [OVN] Wait for the \"Chassis\" creation in the HCG tests\n2806f566f3 Workaround for the OVN/OVS directories in functional job\n4a63fa2092 Change vni_range in openvswitch_type_drivers\nc0f6145b87 [SGL] Ignore port groups that don\u0027t come from SGs\n9751a128a9 [OVN] Use ``uuid.UUID`` to build a OVN database row\n6b6bbd19e7 [eventlet-removal] Handle stop DHCP agent\n5e23d87690 Update IPv6 gateway in OVN after subnet gw changes\n2b63ab6877 [FT] Wait for the manager to be created\n498cd0db13 ruff: Use more specific name to enable pyupgrade rule\n620743fce9 Start running pylint on the test tree\n5b811fcef5 Handle virtual port parent lookup via Neutron DB\n67969c147b Append content-encoding header\n92c7032f76 Initialize DHCP leases with infinite duration\n1c3eac52b7 Remove false-positive ACLs in OVN DB sync\n1b40c5d0ce [doc] Update plugins needed for auto-allocation\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I56e98f7405fb2467173e66c8867bc87f471fe45f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5414dbb0a3fa9c2183a0ff77475196a1f1c3f861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5414dbb0a3fa9c2183a0ff77475196a1f1c3f861"}]},"branch":"refs/heads/master"},"f6f6c1abce56d51ed236d7572c72e6b04372d231":{"kind":"REWORK","_number":2,"created":"2026-01-07 15:52:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/57/972457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/972457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/972457/2"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:48:04.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 15:52:11.000000000","tz":60},"subject":"Release Neutron 28.0.0.0b2","message":"Release Neutron 28.0.0.0b2\n\nThis version contains the unit and functional changes that replace\n\"tenant_id\" by \"project_id\".\n\n$ git log --oneline --no-merges 28.0.0.0b1..af2aefeb87\ne1f95eea4a Fix duplicated checks to enforce usage of assertIs(Not)None\n487552c4de [Docs] Remove note about FWaaS not being supported by OVN backend\nb1419970a5 Change test code to use project_id key\n185b5ffe2b Doc: add warnig for fullstack\nef78d8e978 bgp: Improve existing server Commands\n02b7aac66b Change test code to use project_id in requests\n29aee58292 Change test code to use project_id to create objects\n8bf2b5d318 Change test code to use project_id in dictionaries\nbc096adac1 Change some functional test code to use project_id\n32762f848e Move api-paste.ini, rootwrap.conf files\n680b697247 Change some unit test code to use project_id\n4808ad7901 Remove the \"nftables\" explicit CI jobs\n2faba4ede3 Refactor ovn-db-sync script\nc8d06727c3 bgp: Remove chassis indexing\n4d196d1cd1 Check if network is external in auto-allocate code\n9edf82b24d [OVN] Fix the `_gen_router_port_ext_ids` call\n0e9f221de1 Drop redundant target-version option\n995785b0e8 Remove redundant mock statement\n137949ee07 Search for active port binding when updating LSP host info\nc6b2c1d612 bgp: Introduce Main router and chassis router commands\nd251834cc6 Use NumaAffinityPoliciesEnumField from neutron-lib\n7fe015c871 bgp: Introduce service plugin\nd8ec541122 ruff: Enable E5 check\n28a1483b70 [Docs] Add doc about Virtual IP addresses in Neutron ML2/OVN\na991dd1d6b Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing\n634b457ccf Fix ignored metadata options\n084ffb5866 Remove a missing Windows configuration file\n099e90b2e4 [OVN] OVN agent configuration options list missing \"extensions\"\n75c8ff55ad [OVN] Add documentation/reno for the OVN external port N/S routing\ncdf75bdea7 iPXE over IPv6 supported since OVN v23.06.0\n9cd84180ba [FT] Add extra logs to ``test_create_bridges`` test\n31b78bf23e [FT] Wait for the manager to be created (2)\n64d35a7c12 Change some common test code to use project_id\n6f4917d2d3 [OVN] DNAT registers always have the FIP external MAC\n71ae26e352 Don\u0027t prevent setting CIDRs as allowed_address_pair for port\na924eb9faa [OVN] Add \u0027provider:physical_network\u0027 field if there is physnet\n26513762c1 Allow plugins to add periodics to maintenance worker\n45b8dfe6a8 [OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\n3df3d8ebd2 [eventlet-removal] Deprecate wsgi options\nf2d6eaa61c Bump ovsdbapp to version 2.14.1\nbdcee6d5e0 [docs] Add OVS/OVN min requirement table\n807e6a5cf6 Fix incorrect log message in ovn_client\n27bd4f687b Add `neutron-functional-with-ovsdbapp-master` job\ncdac238f9a Centralize oslo.service patching\n3199846d4c Increase `neutron-functional` job timeout to 2.5 hours\ncf41465164 Use common constant for neutron-ovn-agent process name\n7d908bf182 Add ovn-agent to metadata caching doc\n20585d865d Revert \"[eventlet-removal] Handle stop DHCP agent\"\nd95e3e5eaf Replace response body in after hook for 404 error\n5e92bbfa50 [OVN] Initialize OVN agent in ``start`` method\n5e626f8bda [OVN] Sync the LRP Gateway_Chassis with the network HCG\ne94e7447aa [OVN] The external networks GW chassis must the same as the GW LRP\n78593ce9b8 [OVN] Wait for the \"Chassis\" creation in the HCG tests\n2806f566f3 Workaround for the OVN/OVS directories in functional job\n4a63fa2092 Change vni_range in openvswitch_type_drivers\nc0f6145b87 [SGL] Ignore port groups that don\u0027t come from SGs\n9751a128a9 [OVN] Use ``uuid.UUID`` to build a OVN database row\n6b6bbd19e7 [eventlet-removal] Handle stop DHCP agent\n5e23d87690 Update IPv6 gateway in OVN after subnet gw changes\n2b63ab6877 [FT] Wait for the manager to be created\n498cd0db13 ruff: Use more specific name to enable pyupgrade rule\n620743fce9 Start running pylint on the test tree\n5b811fcef5 Handle virtual port parent lookup via Neutron DB\n67969c147b Append content-encoding header\n92c7032f76 Initialize DHCP leases with infinite duration\n1c3eac52b7 Remove false-positive ACLs in OVN DB sync\n1b40c5d0ce [doc] Update plugins needed for auto-allocation\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I56e98f7405fb2467173e66c8867bc87f471fe45f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6f6c1abce56d51ed236d7572c72e6b04372d231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6f6c1abce56d51ed236d7572c72e6b04372d231"}]},"branch":"refs/heads/master"},"d3983034f4cba04b75d8133a298d933b45c495e2":{"kind":"REWORK","_number":3,"created":"2026-01-08 17:16:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/57/972457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/972457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/972457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/972457/3"}}},"commit":{"parents":[{"commit":"cb624ee51454cf3d2a73c17678dcffef1d1b9577","subject":"Merge \"oslo.serialization 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb624ee51454cf3d2a73c17678dcffef1d1b9577"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-01-07 11:48:04.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-08 17:16:10.000000000","tz":0},"subject":"Release Neutron 28.0.0.0b2","message":"Release Neutron 28.0.0.0b2\n\nThis version contains the unit and functional changes that replace\n\"tenant_id\" by \"project_id\".\n\n$ git log --oneline --no-merges 28.0.0.0b1..644d2b767d\n66eb080852 Add ``path_separator`` to alembic configuration\n77352c3397 Fix list of migration entry points for the ovn-db-sync tool\ne1f95eea4a Fix duplicated checks to enforce usage of assertIs(Not)None\n487552c4de [Docs] Remove note about FWaaS not being supported by OVN backend\nb1419970a5 Change test code to use project_id key\n185b5ffe2b Doc: add warnig for fullstack\nef78d8e978 bgp: Improve existing server Commands\n02b7aac66b Change test code to use project_id in requests\n29aee58292 Change test code to use project_id to create objects\n8bf2b5d318 Change test code to use project_id in dictionaries\nbc096adac1 Change some functional test code to use project_id\n32762f848e Move api-paste.ini, rootwrap.conf files\n680b697247 Change some unit test code to use project_id\n4808ad7901 Remove the \"nftables\" explicit CI jobs\n2faba4ede3 Refactor ovn-db-sync script\nc8d06727c3 bgp: Remove chassis indexing\n4d196d1cd1 Check if network is external in auto-allocate code\n9edf82b24d [OVN] Fix the `_gen_router_port_ext_ids` call\n0e9f221de1 Drop redundant target-version option\n995785b0e8 Remove redundant mock statement\n137949ee07 Search for active port binding when updating LSP host info\nc6b2c1d612 bgp: Introduce Main router and chassis router commands\nd251834cc6 Use NumaAffinityPoliciesEnumField from neutron-lib\n7fe015c871 bgp: Introduce service plugin\nd8ec541122 ruff: Enable E5 check\n28a1483b70 [Docs] Add doc about Virtual IP addresses in Neutron ML2/OVN\na991dd1d6b Fix maintenance LRP prefix adding fails over \u0027name\u0027 attribute missing\n634b457ccf Fix ignored metadata options\n084ffb5866 Remove a missing Windows configuration file\n099e90b2e4 [OVN] OVN agent configuration options list missing \"extensions\"\n75c8ff55ad [OVN] Add documentation/reno for the OVN external port N/S routing\ncdf75bdea7 iPXE over IPv6 supported since OVN v23.06.0\n9cd84180ba [FT] Add extra logs to ``test_create_bridges`` test\n31b78bf23e [FT] Wait for the manager to be created (2)\n64d35a7c12 Change some common test code to use project_id\n6f4917d2d3 [OVN] DNAT registers always have the FIP external MAC\n71ae26e352 Don\u0027t prevent setting CIDRs as allowed_address_pair for port\na924eb9faa [OVN] Add \u0027provider:physical_network\u0027 field if there is physnet\n26513762c1 Allow plugins to add periodics to maintenance worker\n45b8dfe6a8 [OVN] Add \u0027qos-bw-minimum-ingress\u0027 ML2 extension\n3df3d8ebd2 [eventlet-removal] Deprecate wsgi options\nf2d6eaa61c Bump ovsdbapp to version 2.14.1\nbdcee6d5e0 [docs] Add OVS/OVN min requirement table\n807e6a5cf6 Fix incorrect log message in ovn_client\n27bd4f687b Add `neutron-functional-with-ovsdbapp-master` job\ncdac238f9a Centralize oslo.service patching\n3199846d4c Increase `neutron-functional` job timeout to 2.5 hours\ncf41465164 Use common constant for neutron-ovn-agent process name\n7d908bf182 Add ovn-agent to metadata caching doc\n20585d865d Revert \"[eventlet-removal] Handle stop DHCP agent\"\nd95e3e5eaf Replace response body in after hook for 404 error\n5e92bbfa50 [OVN] Initialize OVN agent in ``start`` method\n5e626f8bda [OVN] Sync the LRP Gateway_Chassis with the network HCG\ne94e7447aa [OVN] The external networks GW chassis must the same as the GW LRP\n78593ce9b8 [OVN] Wait for the \"Chassis\" creation in the HCG tests\n2806f566f3 Workaround for the OVN/OVS directories in functional job\n4a63fa2092 Change vni_range in openvswitch_type_drivers\nc0f6145b87 [SGL] Ignore port groups that don\u0027t come from SGs\n9751a128a9 [OVN] Use ``uuid.UUID`` to build a OVN database row\n6b6bbd19e7 [eventlet-removal] Handle stop DHCP agent\n5e23d87690 Update IPv6 gateway in OVN after subnet gw changes\n2b63ab6877 [FT] Wait for the manager to be created\n498cd0db13 ruff: Use more specific name to enable pyupgrade rule\n620743fce9 Start running pylint on the test tree\n5b811fcef5 Handle virtual port parent lookup via Neutron DB\n67969c147b Append content-encoding header\n92c7032f76 Initialize DHCP leases with infinite duration\n1c3eac52b7 Remove false-positive ACLs in OVN DB sync\n1b40c5d0ce [doc] Update plugins needed for auto-allocation\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I56e98f7405fb2467173e66c8867bc87f471fe45f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3983034f4cba04b75d8133a298d933b45c495e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3983034f4cba04b75d8133a298d933b45c495e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
