)]}'
{"id":"openstack%2Freleases~974034","triplet_id":"openstack%2Freleases~master~I160958d554e670b81c40a837a3ed9a32712c844e","project":"openstack/releases","branch":"master","topic":"ironicclient-gazpacho-5150","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-01-22 16:00:27.000000000","reason":"\u003cGERRIT_ACCOUNT_5890\u003e replied on the change","reason_account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-01-26 13:24:49.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2026-01-26 13:24:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I160958d554e670b81c40a837a3ed9a32712c844e","subject":"Release ironicclient 5.15.0 for gazpacho","status":"MERGED","created":"2026-01-21 10:33:24.000000000","updated":"2026-01-26 13:24:49.000000000","submitted":"2026-01-26 13:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974034-ironicclient-gazpacho-5150","meta_rev_id":"802ff2de4c10f0f245983cf465b9f18e90d220d1","_number":974034,"virtual_id_number":974034,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 13:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-01-22 16:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-01-23 10:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-01-26 13:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-01-22 15:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 10:34:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 12:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2026-01-21 14:42:01.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-01-23 10:01:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"6c3940f92e1f6e601dc495ce6f28d43758da9f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-21 10:33:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acede1c9c9c670a6f904727a1098f80258318f73","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:34:03.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"26a2a3778a0463715f3f5b5e796cf5e9093013c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 10:38:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2545108bf629421091acf35352313647\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eadac471eb34444b317342de31798a9 : SUCCESS in 4m 07s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/73c176ebea1f4929ba0eff4fa7fa78a1 : SUCCESS in 3m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3e963a84e2124040a8673d462bacf1dc : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fd399f43003c79205709a26215b396e105b221be","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-21 14:42:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a4cd8f74376275dfa5c0e05b1822bc388eda5f85","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-21 15:00:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21c39230de91fb3e47248fc2c943ff497219676f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-21 15:02:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5d775cff803616cbdf948e5f6bf6a9d84c047856","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-21 15:03:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ca5e7b6b03d5bd02366dd896b2b8d21ff7d335a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-21 15:50:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f5daab4ce7b9c3dabe4806c7b760790bc6ff5d8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 15:56:07.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"aeabb8d52a4118db28a6bb37879e8560de18ba02","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-21 15:59:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55464bbcc6b3322f27389d09e841a3910610ded8","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-21 16:03:16.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"f47d4dbb689f2f6e2924e05274f2f25a45f4a457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 16:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52153ae32bab40b385612656aa82e02d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c03f5d3947f458abfddba66a446dd73 : SUCCESS in 8m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/405d73e9bcae48608e5c8ad0637792d3 : SUCCESS in 6m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/32107671a768497abc5a2cfdec2f5dae : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"71bd3d06475a6065c2fc4bbc85735220f3b491fc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-21 16:53:58.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8d197a6a748aadd4d4b00c7ede14190b63661e6c","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-21 17:03:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df0eb58ab7c47b1c38f24320ba1107b878adf753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-22 15:56:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"89f985a968fea7efae58b6f3e9fc76be9cdcf870","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:58:18.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a4af06ace17d411820d5f9513fab5829b44dc697","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-01-22 16:00:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"add8099055b0390d1b4c12f887fe5f4de984a3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:08:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e35309c3ac47deb6a80bc1ce661fc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7854c1e1cfbb4e07a8c1991b8ddaa124 : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/0240e28ff1e441158db011df3fed8e81 : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b6f81b8e307d4abe9ec96ed2e0e27e44 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"32b39385c730bf9c34a167dc9fa9a10f074488bc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-23 10:01:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"177f846a4a2cec4b6f33ac7d0d22a114528ea1d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-26 13:12:27.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"139ae679a0ec6f0f14c0bd8ff9302614eb18bb06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:12:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc256929abb87d708bff13c9bf55c73a390fd4e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:24:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5cd4ab16a86467bb0d12308081e3148\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f769a49a654dfb87ba57aa917564be : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cdc19c14df4366a3b3daa5271bebf0 : SUCCESS in 6m 24s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/25f9a273429f4f61ad1ec4b2557716c3 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"802ff2de4c10f0f245983cf465b9f18e90d220d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 13:24:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3375bc44801b54f4f54a7c06a3e3bb58ba255ec9","revisions":{"370dac61feb3040aa874c9db10dd5a7d36ee2d40":{"kind":"REWORK","_number":1,"created":"2026-01-21 10:33:24.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/34/974034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/974034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/974034/1"}}},"commit":{"parents":[{"commit":"84ed370c0b01a81dbe7298935fdc0fe7a6340825","subject":"Merge \"Release keystonemiddleware stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84ed370c0b01a81dbe7298935fdc0fe7a6340825"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-21 10:33:05.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-21 10:33:20.000000000","tz":60},"subject":"Release ironicclient 5.15.0 for gazpacho","message":"Release ironicclient 5.15.0 for gazpacho\n\nChange-Id: I160958d554e670b81c40a837a3ed9a32712c844e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/370dac61feb3040aa874c9db10dd5a7d36ee2d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/370dac61feb3040aa874c9db10dd5a7d36ee2d40"}]},"branch":"refs/heads/master"},"9df90811d8b4e8a3c0c535e11182adb9e000db9d":{"kind":"REWORK","_number":2,"created":"2026-01-21 15:50:52.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/34/974034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/974034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/974034/2"}}},"commit":{"parents":[{"commit":"84ed370c0b01a81dbe7298935fdc0fe7a6340825","subject":"Merge \"Release keystonemiddleware stable/2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84ed370c0b01a81dbe7298935fdc0fe7a6340825"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-21 10:33:05.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-21 15:50:48.000000000","tz":60},"subject":"Release ironicclient 5.15.0 for gazpacho","message":"Release ironicclient 5.15.0 for gazpacho\n\nChange-Id: I160958d554e670b81c40a837a3ed9a32712c844e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9df90811d8b4e8a3c0c535e11182adb9e000db9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9df90811d8b4e8a3c0c535e11182adb9e000db9d"}]},"branch":"refs/heads/master"},"3375bc44801b54f4f54a7c06a3e3bb58ba255ec9":{"kind":"REWORK","_number":3,"created":"2026-01-22 15:56:59.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/34/974034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/974034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/974034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/974034/3"}}},"commit":{"parents":[{"commit":"1ae5e0b93570577ff736d5e2cd6307bb29664459","subject":"Merge \"Replace obsolete PCRE packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ae5e0b93570577ff736d5e2cd6307bb29664459"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-21 10:33:05.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-01-22 15:56:51.000000000","tz":60},"subject":"Release ironicclient 5.15.0 for gazpacho","message":"Release ironicclient 5.15.0 for gazpacho\n\nChange-Id: I160958d554e670b81c40a837a3ed9a32712c844e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3375bc44801b54f4f54a7c06a3e3bb58ba255ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3375bc44801b54f4f54a7c06a3e3bb58ba255ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
