)]}'
{"id":"openstack%2Freleases~976625","triplet_id":"openstack%2Freleases~master~I97aa3c85ce21ca6323b2490c46339afcc61ce353","project":"openstack/releases","branch":"master","topic":"remove-pkg_resources","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-03-10 12:11:36.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-03-10 12:11:36.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-10 12:11:36.000000000","reason":"Change was abandoned"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2026-03-10 12:11:36.000000000","reason":"Change was abandoned"}},"hashtags":["remove-pkg_resources"],"change_id":"I97aa3c85ce21ca6323b2490c46339afcc61ce353","subject":"Normalize xstatic package configurations","status":"ABANDONED","created":"2026-02-12 14:05:13.000000000","updated":"2026-03-10 12:11:36.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83b7fc280a37bebcb3ed8a4f9f3ba12f7f09d04c","_number":976625,"virtual_id_number":976625,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-25 14:50:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 14:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-12 14:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-02-12 14:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2026-02-12 15:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 15:34:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-02-18 20:45:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a1c827bfc40ca0222e04148c3f142906145d0283","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 14:05:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1375735f3c8f8ac2e24f7b3edae0e99c48c06efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 15:01:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72b1aba49d179b004a2b38128b85c5c8c8f8671b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:29:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73078fe508aa4570b7596423999bfd8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a73b8be4c77418a8fee9538d1d6528d : SUCCESS in 6m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/99f075deb313437a9d71c303c762227f : SUCCESS in 18m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c2b2049f4b144db983cdedcbd210e8dd : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"6af1741cdb3b214bfad1e8fcd4caafc8b299d61d","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-12 15:43:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1e8e3bf0557f98ae7f88cec893999f1212c9ab5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-12 18:09:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"32d4bf2df9202a4743f0bf1ef3588de6f6c92136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:16:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22d94d734c9543668121793fb565286c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c08d3fc2b641b8b953589a141d0d05 : SUCCESS in 6m 27s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/02539fe7062d401e8bbf11d600f3a928 : SUCCESS in 17m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5e271583a803411481a547b521e242b4 : TIMED_OUT in 1h 01m 05s","accounts_in_message":[],"_revision_number":3},{"id":"33bfc3aef0de845bfc677c7be31850f733add3f8","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-12 21:21:55.000000000","message":"Hashtag added: remove-pkg_resources","accounts_in_message":[],"_revision_number":3},{"id":"1de9b79e11edfca26f6b93b7eccb75add9da47ce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-13 16:03:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fe40ce2da47b061dfb78d736ca73cc9d261f6db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:27:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed635af4b314367beb9cd59a57bc733\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0af7195e7b141eca0f11c5135a86a27 : SUCCESS in 6m 48s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/aaf38bed1c6c4680b2b9e2f22ac43d81 : SUCCESS in 8m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fa3b806987e94b6da3f5a00c67ab55e2 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":3},{"id":"bf2042235ce808cb09c8012a82ad4dd68e4de3ea","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-13 18:35:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"476a1a3d420c2946d4931580422bb6ab9685bd55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-13 18:47:34.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6bb87fc4e8ec47425765e55df60b85e0604f7cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 18:54:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87cc6de14d814c3aab40f018faa7193f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb95ce2729e49a8aa2c3bd880b32adf : SUCCESS in 6m 03s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7f34fab2617d47f0bfe310fc9882d481 : SUCCESS in 3m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/80342c65a51147209d4671a38db71921 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"c05d98df3949c9f17bee97a27db56c2986bdfb7e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-16 15:34:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c1bcf1dc90f68431e707d603ec01e6e09fa2470a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-18 10:17:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d8f9fa1c2446e48d6bf2dd83c9d91b2775f1e532","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-18 10:17:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3452190dc1f32bb2fab3c6a11833012526caead5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 10:24:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8587c6868ef2427b9b69dc8e15c1cf39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d779f356a14d11a34be701fb684006 : SUCCESS in 5m 53s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/717eb9aa42a54ddf9284d1627aa4496c : SUCCESS in 3m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/15db92effd03476fbd47f71e6d14b379 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"c28aec66f4969e441def190a9833bf82973f1db8","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-18 17:16:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"92e99fb112b795dd24b9c4d1c4196ccca0f0479d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-18 20:45:23.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"835b1cf1be47d0fa70af57d77a74ac8a7a7ba09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-25 14:34:01.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"94596f2a272e9f2cc5599c2dcde7f405185f0200","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-25 14:34:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"447cf3047ba5302249b201ceff304bc1e4caff31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 14:50:05.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c280a1f58a634e2b9a3e3246da6144d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8cfb92134b04895ade7a480877c0128 : SUCCESS in 7m 53s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/005f876ddadb4e13af52000db24a6208 : FAILURE in 4m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/df6c59ca01ba4772a820cdc8c4b721f6 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":6},{"id":"83b7fc280a37bebcb3ed8a4f9f3ba12f7f09d04c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-10 12:11:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4580bdee361d72fe5da2bdfdaec0a633ffd1d79f","revisions":{"820ed6b8f09a831e44156282ba1da5c6faedfa9b":{"kind":"REWORK","_number":1,"created":"2026-02-12 14:05:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/1"}}},"commit":{"parents":[{"commit":"0c52ce6e8f32a0cc4de7a5b15c3be8dd3505ec29","subject":"Bulk release newly onboarded xstatic packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c52ce6e8f32a0cc4de7a5b15c3be8dd3505ec29"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:05:04.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/820ed6b8f09a831e44156282ba1da5c6faedfa9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/820ed6b8f09a831e44156282ba1da5c6faedfa9b"}]},"branch":"refs/heads/master"},"d7751137cf3917c80884b66154e84d41626509ef":{"kind":"REWORK","_number":2,"created":"2026-02-12 15:01:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/2"}}},"commit":{"parents":[{"commit":"1c4cc6d48b70f019ed8dd6cf57c615559b896da8","subject":"Bulk release newly onboarded xstatic packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c4cc6d48b70f019ed8dd6cf57c615559b896da8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 15:00:56.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7751137cf3917c80884b66154e84d41626509ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7751137cf3917c80884b66154e84d41626509ef"}]},"branch":"refs/heads/master"},"cd9c4447001acb04e10cbbfb49d109fc37c6a4b7":{"kind":"REWORK","_number":3,"created":"2026-02-12 18:09:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/3"}}},"commit":{"parents":[{"commit":"d8bcbd3eab45e64bdc7e1615b99d8eeb43b039bb","subject":"Bulk release newly onboarded xstatic packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d8bcbd3eab45e64bdc7e1615b99d8eeb43b039bb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 18:08:19.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd9c4447001acb04e10cbbfb49d109fc37c6a4b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd9c4447001acb04e10cbbfb49d109fc37c6a4b7"}]},"branch":"refs/heads/master"},"a22b2c2081a98c1680e96d6b7e9df34de99752af":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-13 18:47:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/4"}}},"commit":{"parents":[{"commit":"9ccbf55d73a1557394d8ba699f054cdfd39d864d","subject":"Bulk release newly onboarded xstatic packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9ccbf55d73a1557394d8ba699f054cdfd39d864d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-13 18:47:34.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a22b2c2081a98c1680e96d6b7e9df34de99752af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a22b2c2081a98c1680e96d6b7e9df34de99752af"}]},"branch":"refs/heads/master","description":"Rebase"},"ad30481593117e06de6b54d6ca63686c55659755":{"kind":"REWORK","_number":5,"created":"2026-02-18 10:17:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/5"}}},"commit":{"parents":[{"commit":"b3617175b7eb55adb0facdd0494cc27a59068611","subject":"Merge \"Release final taskflow for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3617175b7eb55adb0facdd0494cc27a59068611"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-18 10:02:10.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ad30481593117e06de6b54d6ca63686c55659755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ad30481593117e06de6b54d6ca63686c55659755"}]},"branch":"refs/heads/master"},"4580bdee361d72fe5da2bdfdaec0a633ffd1d79f":{"kind":"REWORK","_number":6,"created":"2026-02-25 14:34:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/25/976625/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/976625/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/976625/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/976625/6"}}},"commit":{"parents":[{"commit":"b3617175b7eb55adb0facdd0494cc27a59068611","subject":"Merge \"Release final taskflow for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b3617175b7eb55adb0facdd0494cc27a59068611"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-12 14:03:04.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-25 14:34:00.000000000","tz":0},"subject":"Normalize xstatic package configurations","message":"Normalize xstatic package configurations\n\nTo ease future bulk updates.\n\nChange-Id: I97aa3c85ce21ca6323b2490c46339afcc61ce353\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4580bdee361d72fe5da2bdfdaec0a633ffd1d79f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4580bdee361d72fe5da2bdfdaec0a633ffd1d79f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
