)]}'
{"id":"openstack%2Freleases~976799","triplet_id":"openstack%2Freleases~master~I46b9e328f1d905b04a1bbcb60c3b5831363ad777","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-02-16 09:41:24.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-02-16 09:41:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46b9e328f1d905b04a1bbcb60c3b5831363ad777","subject":"[manila] New 2025.1 Epoxy release","status":"MERGED","created":"2026-02-13 18:01:14.000000000","updated":"2026-02-16 09:41:24.000000000","submitted":"2026-02-16 09:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976799","meta_rev_id":"512a1ab23b430181d231979ae3cd3887d5c357bc","_number":976799,"virtual_id_number":976799,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 09:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-02-16 09:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-02-16 09:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-02-13 18:14:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 18:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 18:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-02-16 09:27:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"def7d618599cfa20d2c48ead80e38865f821be0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-13 18:01:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85e8b2b66d527afe0535bd5106599c0cf97ada66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-13 18:01:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1a4468d5fe589ec1dbdc17ed223aaf3ccb1df1e8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 18:02:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"1741bbcd1f76e038e22ec9b17f5d62fa20d71345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-02-13 18:09:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":3},{"id":"47173fdb3b600ad745126597dac6b6a768af2faa","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 18:14:09.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"411cbd817d89f0e1fdec4887fd29cd64574ddc70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 18:28:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b7acc364ce4c148cbcecd12529116a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0b19b7ab1c4e5591e8e091b43bfd49 : SUCCESS in 7m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e706bcd40119409f8c2f1fb7979d806e : SUCCESS in 3m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/627005bc33e846b98d12a8045d08ba97 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"fb41e5602d1e567ccc086187b05be524183efe9a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-16 09:27:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f5510225984678e61a824d6a399c0d1278d509d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 09:27:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5702e856fb2598daf3304c9c7daab5ad669ae2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 09:41:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81bd3968976548d18ab94e0b4dd315c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70126b22742473d812fe3ddd8329b99 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9f5124c0ea4496a54fed76cfd404bc : SUCCESS in 7m 07s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/61c2f21c1284430ebaa4f387508ff7c9 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"512a1ab23b430181d231979ae3cd3887d5c357bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 09:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54463625ebce72390a144f0e8540a33e8a7afe06","revisions":{"77f3fdca104ba859163c9281192dbc0dca5da3f9":{"kind":"REWORK","_number":1,"created":"2026-02-13 18:01:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/976799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/976799/1"}}},"commit":{"parents":[{"commit":"8aa731fc49a4612167bcc79bb8cf05e999319fe2","subject":"[manila] Release 2025.2 stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8aa731fc49a4612167bcc79bb8cf05e999319fe2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:00:30.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:00:30.000000000","tz":-180},"subject":"[manila] New 2025.1 Epoxy release","message":"[manila] New 2025.1 Epoxy release\n\nChange-Id: I46b9e328f1d905b04a1bbcb60c3b5831363ad777\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77f3fdca104ba859163c9281192dbc0dca5da3f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77f3fdca104ba859163c9281192dbc0dca5da3f9"}]},"branch":"refs/heads/master"},"bfdaa90b4c31ba143e184986e452f3232411d5fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-13 18:01:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/976799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/976799/2"}}},"commit":{"parents":[{"commit":"8aa731fc49a4612167bcc79bb8cf05e999319fe2","subject":"[manila] Release 2025.2 stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8aa731fc49a4612167bcc79bb8cf05e999319fe2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:00:30.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:01:29.000000000","tz":-180},"subject":"[manila] New 2025.1 Epoxy release","message":"[manila] New 2025.1 Epoxy release\n\nChanges include:\n$ git log --oneline --no-merges 20.0.0..066a6d9\n066a6d9f5 (origin/stable/2025.1, gerrit/stable/2025.1) bug/2130717 Fix KeyError in modify_volume when autosize attributes missing\n908cc91ce Fix shares getting stuck in ensuring status\n721759410 Allow configuring netapp volume retention period to zero\n0d87f4b8f [NetApp] guard already existing cifs access\n3904baecc Improve scheduler performance on thin provisioning\n7c7fb884b Dell Unity: Fix driver startup issue with LACP configured\nb57371e57 Handle Neuton Subnet Full Exception\n36b202934 Fix ensure shares operation\ncfc8ccef6 [NetApp-REST] Fixing basic share creation failure\n9994dde8a Fix server instance subnet cleanup\nf957732fa Update get_snapshot to return 404 if snapshot is inaccessible\n40a4fd565 NetApp: Fix vserver create REST API\n744bbbb44 NetApp: Fix export location during replica promote\n47c9a6ebf improve db calls in periodic replica tasks\nb4168f42d Add pyproject.toml\neaf4c0a15 Add new manila.wsgi module\ne055c6583 Handle race condition for share server delete\n9d1bede74 Fix \u0027server_migrating\u0027 status of non-active replica\n\nChange-Id: I46b9e328f1d905b04a1bbcb60c3b5831363ad777\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfdaa90b4c31ba143e184986e452f3232411d5fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfdaa90b4c31ba143e184986e452f3232411d5fa"}]},"branch":"refs/heads/master"},"54463625ebce72390a144f0e8540a33e8a7afe06":{"kind":"REWORK","_number":3,"created":"2026-02-13 18:09:38.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/976799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/976799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/976799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/976799/3"}}},"commit":{"parents":[{"commit":"8aa731fc49a4612167bcc79bb8cf05e999319fe2","subject":"[manila] Release 2025.2 stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8aa731fc49a4612167bcc79bb8cf05e999319fe2"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:00:30.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-02-13 18:09:27.000000000","tz":-180},"subject":"[manila] New 2025.1 Epoxy release","message":"[manila] New 2025.1 Epoxy release\n\nChanges include:\n$ git log --oneline --no-merges 20.0.0..066a6d9\n066a6d9f5 (origin/stable/2025.1, gerrit/stable/2025.1) bug/2130717 Fix KeyError in modify_volume when autosize attributes missing\n908cc91ce Fix shares getting stuck in ensuring status\n721759410 Allow configuring netapp volume retention period to zero\n0d87f4b8f [NetApp] guard already existing cifs access\n3904baecc Improve scheduler performance on thin provisioning\n7c7fb884b Dell Unity: Fix driver startup issue with LACP configured\nb57371e57 Handle Neuton Subnet Full Exception\n36b202934 Fix ensure shares operation\ncfc8ccef6 [NetApp-REST] Fixing basic share creation failure\n9994dde8a Fix server instance subnet cleanup\nf957732fa Update get_snapshot to return 404 if snapshot is inaccessible\n40a4fd565 NetApp: Fix vserver create REST API\n744bbbb44 NetApp: Fix export location during replica promote\n47c9a6ebf improve db calls in periodic replica tasks\nb4168f42d Add pyproject.toml\neaf4c0a15 Add new manila.wsgi module\ne055c6583 Handle race condition for share server delete\n9d1bede74 Fix \u0027server_migrating\u0027 status of non-active replica\n\nChange-Id: I46b9e328f1d905b04a1bbcb60c3b5831363ad777\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54463625ebce72390a144f0e8540a33e8a7afe06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/54463625ebce72390a144f0e8540a33e8a7afe06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
