)]}'
{"id":"openstack%2Freleases~978412","triplet_id":"openstack%2Freleases~master~I485b070d7ea11c5045eb04dd4e6da7cfd5c954fe","project":"openstack/releases","branch":"master","topic":"gazpacho-stable-branches","attention_set":{},"removed_from_attention_set":{"13940":{"account":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"last_update":"2026-03-06 13:12:01.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-06 13:12:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I485b070d7ea11c5045eb04dd4e6da7cfd5c954fe","subject":"[freezer] Create 2026.1 branch for client and non-client libs","status":"MERGED","created":"2026-03-02 10:05:24.000000000","updated":"2026-03-06 13:12:01.000000000","submitted":"2026-03-06 13:12:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978412-gazpacho-stable-branches","meta_rev_id":"d4e57f0d39a13121468453d13f939c381aacea5e","_number":978412,"virtual_id_number":978412,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 13:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-03-06 13:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-03-06 13:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-03-06 13:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 10:19:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-02 10:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 12:20:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-03-02 12:20:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"}],"messages":[{"id":"9c6c49aa994b922d5c06b3686853f8142210de4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-02 10:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38aeee3e8d1cf49bc574166f3ffed4c99abc1b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 10:33:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e8c8897a2e46deb66d58aa35994af9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/befb9ce81c0c47b5888b397da9fb59f0 : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b5bc6ef840d34db183fca83468cd667e : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7307bedda5ec43888613eb79eeb3a20f : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4ca1416f225feee063063b70911eaa7e8c54a979","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-06 13:00:03.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b450126b9f2afa4b0899243a1e4d09930d0f9c73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 13:00:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4e428ee33a305cbc44efa95ef2ecad02274cccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 13:12:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f391867c0c4503a56ca1bdd8bcbe98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ff23fef70a4990b593afe28ea4d9a8 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe44d9bd03f4b4294312e1702889213 : SUCCESS in 7m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/374b3db60c4449d18471243039d71f8c : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d4e57f0d39a13121468453d13f939c381aacea5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 13:12:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ecd006d76b0e3e5f5129a70ee8cc3e5cc12d4b2","revisions":{"2ecd006d76b0e3e5f5129a70ee8cc3e5cc12d4b2":{"kind":"REWORK","_number":1,"created":"2026-03-02 10:05:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/12/978412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/978412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/978412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/978412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/978412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/978412/1"}}},"commit":{"parents":[{"commit":"01a154a0e9dd52ab2b1156f4b43fbf74beeda65e","subject":"Merge \"Cut stable/2026.1 for os-brick\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01a154a0e9dd52ab2b1156f4b43fbf74beeda65e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-02 10:01:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-02 10:01:31.000000000","tz":60},"subject":"[freezer] Create 2026.1 branch for client and non-client libs","message":"[freezer] Create 2026.1 branch for client and non-client libs\n\nThis creates the stable/2026.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch gets no response, then this will be merged\nafter Thursday, March 5th.\n\nChange-Id: I485b070d7ea11c5045eb04dd4e6da7cfd5c954fe\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ecd006d76b0e3e5f5129a70ee8cc3e5cc12d4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ecd006d76b0e3e5f5129a70ee8cc3e5cc12d4b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
