)]}'
{"id":"openstack%2Freleases~979547","triplet_id":"openstack%2Freleases~master~I2e01cb57ad5618233922d56f25e3451557f9ad2b","project":"openstack/releases","branch":"master","topic":"gazpacho-rc1-deadline","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-03-13 07:11:34.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-03-13 07:11:34.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-03-13 07:06:41.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I2e01cb57ad5618233922d56f25e3451557f9ad2b","subject":"Release manila RC1 for 2026.1 Gazpacho","status":"MERGED","created":"2026-03-09 12:14:19.000000000","updated":"2026-03-13 07:11:34.000000000","submitted":"2026-03-13 07:11:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979547-gazpacho-rc1-deadline","meta_rev_id":"c83ccfffb472bfca8bf434ed1b331314f2e82911","_number":979547,"virtual_id_number":979547,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 07:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2026-03-12 23:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2026-03-13 07:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2026-03-13 07:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-03-12 23:32:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 12:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-09 12:21:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-03-09 12:21:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-03-09 12:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b405013d2dcbe4cb489db059d50f127fb4b95e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-09 12:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10afd557cdb08ecedc44aba0e50406094820006c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 12:28:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd49dd010ff64785b0d1e56ecf586944\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa220a70a424d759494c41cba408ecb : SUCCESS in 6m 46s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/dc69184810764763bd10dbd50f1a2025 : SUCCESS in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d6b10e4475844bc6add91e15a9e11142 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"59fa7410a9d3a29cfee4683f402664fbe32bb24b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-03-09 13:00:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8310271fa1f5d373d90aea001f809b26cb4f7904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-12 23:31:30.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8a9707e684d9b6070822989a24a215f28bbd7377","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-03-12 23:32:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21d4d65d8fb67ff54f1ff383a1dacc37d7a8c9cc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 23:32:15.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"6aca6ba8e72bd5f5764293093492ca29e0438b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 23:41:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3d68e5803d45599b302dbc949b431a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3d405a70074ec4b40437b1c0e2f365 : SUCCESS in 4m 25s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/51e5d41cb64d475889a23757e8181fa2 : SUCCESS in 2m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bb3b87b16809492498648b0f33520cd6 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4db128e39ed07e1ddc411c4555a4001e2f31651f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-13 07:06:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d7c4d9599e16de8a324a1ff1393e9ca8305fbb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 07:06:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1978f4ab37316be0407e7658f1439e72e1b7fdf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 07:11:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9f3cf4c364463ca10f450d8d3e5836\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f96b9e8d32473383c1f0e08f778a58 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2487673769f408a9843bee6de4bb9d6 : SUCCESS in 3m 34s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7011a4391f454ab8963d286dd633155f : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c83ccfffb472bfca8bf434ed1b331314f2e82911","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 07:11:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1c00477d2692a0bb8a9cceadcea841a1ff92383","revisions":{"77376a0ec6b6c462988e41c2ab06fa4ef254d650":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:14:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/979547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/979547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/979547/1"}}},"commit":{"parents":[{"commit":"c809854719e22beba4fe604e131351e6387ce3ed","subject":"Merge \"[manila] Gazpacho cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c809854719e22beba4fe604e131351e6387ce3ed"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 12:14:13.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 12:14:13.000000000","tz":60},"subject":"Release manila RC1 for 2026.1 Gazpacho","message":"Release manila RC1 for 2026.1 Gazpacho\n\nThis proposed a RC1 release for manila to get any merged changes\nout prior to the RC1 deadline for 2026.1 Gazpacho going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready. If there is nothing\nto wait for, then please do respond on this patch with +1 as soon as\npossible.\n\nIn any case, if the patch gets no response, then this will be merged\non Friday, March 13th.\n\nChange-Id: I2e01cb57ad5618233922d56f25e3451557f9ad2b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77376a0ec6b6c462988e41c2ab06fa4ef254d650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77376a0ec6b6c462988e41c2ab06fa4ef254d650"}]},"branch":"refs/heads/master"},"b1c00477d2692a0bb8a9cceadcea841a1ff92383":{"kind":"REWORK","_number":2,"created":"2026-03-12 23:31:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/47/979547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/979547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/979547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/979547/2"}}},"commit":{"parents":[{"commit":"c809854719e22beba4fe604e131351e6387ce3ed","subject":"Merge \"[manila] Gazpacho cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c809854719e22beba4fe604e131351e6387ce3ed"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-09 12:14:13.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-03-12 23:31:30.000000000","tz":0},"subject":"Release manila RC1 for 2026.1 Gazpacho","message":"Release manila RC1 for 2026.1 Gazpacho\n\n$ git log 21.0.1..master --no-merges --oneline\n931a76ff5 Dell PowerScale: Fix CIFS manage path resolution and snapshot path\na244262d8 Fix generic driver NFS share deletion \u0027target is busy\u0027 error\nbe73486da Fix grenade skip level job\n6536093d0 Follow up on share replica metadata changes\nd60b2a72a Fix RADOS export index dangling reference in remove_export()\n9b4303f42 Fix share group snapshots with CephFS\nff08d9802 NetApp: Fix all_squash by setting user\n2d3773e44 Fix volume reserved space issue by adding tune2fs\nb398ffdd7 [CephFS driver] Make share deletion idempotent\nbf3c835c9 Remove privileged fields from networks APIs\n4a85dd851 Deprecate the built-in eventlet WSGI server\n0dcbe7123 Initialize context properly so it can be serialized\n93d64d74d CI: Add unit test job with native threading\n6ab2a1e59 Fix pagination links with incorrect endpoint names\n953dec811 NetApp: Mock _start_periodic_tasks in setup error tests\n850295b78 Netapp: cifs - enable SMB signing\nd46c92184 modify lazy loading for ShareInstance model\n1379a19de Fix QoS type review optimizations.\n5cd4fef55 Fix Service.stop() to not hang with threading backend\n1d9352526 Replace eventlet GreenPool\n80971097f Replace greenthread.sleep with time.sleep\n37217c36f Fix delete scheduler hints when share not found\n8c4deea2a Allow running manila services in threading mode\n598485a30 NetApp: Update created_at during manage operations\n76dcba7a1 Fix potential double join in export locs query\n6df5b4208 decorate wrapper method with functools.wrap to keep func name in hooks\nc10154bc7 Adopt to new castellan\n2c193c4ec NetApp-Fix Sync replication related feedback comments\n096b9a680 unused share servers can be transitioning\nfb7790687 Parse and validate share protocol options early\n5019be095 PowerScale documentation update\n2b512a4e1 Implementation of NAE support for NetApp Driver\n5c290b901 Netapp: Add support for QoS type specs\n1bc797a7e netapp: fix pylint errors\nbfcc76a3d Fix pylint issues from 973777\na4f509e32 Metadata for Share Replica Resource\n0aa46de4b HPE: Lint and Documentation fixes for Alletra MP Driver\n2c1c36b4f [NetApp] Fix mount format while managing share\ne8314ea43 Add support for QoS type and specs\n7d85b89c4 NetApp - Add support for Synchronous replication for share\n9fa2ba008 Follow up on API v1 removal\n0e6e4f61e Allow services API to filter by ensuring\nc66a1bd0c Validate auth_strategy parameter\n70d9ba909 Imported Translations from Zanata\nfa13cec40 Allow using mount_point_name when managing share\na329b5125 devstack: Use common function to manage keystoneauth options\nf0685679e HPE: Add new share driver for HPE Alletra MP B10000 storage\n8e0bbbcae api: Remove use of mixin classes\n8b94409fb api: Remove v1 API\n678a3ebb8 Dell PowerScale: resolve managed share path in _get_container_path\ne1fb16f88 tests: Remove tests for v1 API\nb0cf36fed Dell PowerScale: Follow up patch for 972278\n5043ee880 Dell PowerScale: Dedupe support at share level\n9b806a511 docs: Remove references to v1 API\n3b923dd7c api: Remove references to v1\n6fdae0dc0 api: Move v1 controllers to v2 (2/2)\n7262521ff Dell PowerScale: Added support of mount point name\n9e8efd597 Dell PowerScale: Addded support of snapshot mount\na4c93dd31 Dell PowerScale: Added support of manage/unmanage share and snapshot\ned4b05c51 [doc] Warn admins about share type mutability\n574fc1bde pre-commit: Bump versions\n5a707a7b8 api: Remove unnecessary aliasing/brackets\n1cea1c65b api: Fix schema for quota class sets\nfe05f588f Support BFV for DHSS\n5379f99c7 PowerScale Manila: Add share shrink support\na5e675c38 Add schema validation for ensure-shares endpoint\n8140c1f48 Add description to revert schema paramater\nf8bd1b0ec NetApp: Delete parent snapshot from child share after clone\n6e260f46a api: Add Schemas for share instance export locations\n811be781d api: Move v1 controllers to v2 (1/2)\nae3428b18 [Netapp] Add support for prune deleted volumes\n74665efe4 NetApp - Implemented volume autosize reset support for REST\ne28a519bc Adding JSON Schema to Quota Class Sets\na9de9e355 [doc] Add pre-commit instructions\n3530f4ab9 Drop redundant hacking check for xrange\nb6243320e Remove old hacking check for oslo namespace package\nbfdbe0c50 improve listing share servers with search option\nd3f5a2d22 Use oslo.utils implementation to parse/escape server address\n9df958da0 Drop reference to [DEFAULT] secure_proxy_ssl_header\n61888cc21 Run RBAC job in check/gate\ncef2bd8af Fix schema for all_tenants query parameter\n4c07d8238 Validate enabled_share_protocols at confing parsing layer\ncd80e93e2 NetApp: Add min-throughput for QoS policy group create/modify\n728244408 Drop redundant overrides\n63c3ec369 Fix domain ids lost during rpc call\n278992d14 api: Add schema for share revert action\nf83cfbe15 api: Add schema for share migration_start, migration_get_progress actions\ncb932f021 api: Add schema for share migration_complete, migration_cancel actions\n3c0e89cf0 api: Add schema for share force_delete action\n10a9d4e04 api: Add schema for share reset_status, reset_task_state actions\n75e895d57 Fix rally jobs\ne153f284a Update show share export location to return 404 error when the location does not belong to the share\n9fe4a0485 Netapp - Enable all_squash for nfs\n9f2d741db [DOC] Add export location metadata documentation\ne39ddae32 Fix cross-AZ share creation from snapshot\nefa75272f doc: Update Snapshot list and detail API reference to match schema validation\ne07ea2903 Add schema validation for Share Snapshot list API\n1a182180a fix netapp delete_share private_storage cleanup\n8c5874ac7 Reduce NetApp unit test runtime: time.sleep\n4dc5390ed Fix KeyError in modify_volume when autosize attributes missing\n122d96d89 Allow configuring netapp volume retention period to zero\nd434adbc3 grenade: Start/stop services using unit names\n58e8f893e Edit deferred_deletion doc\nea32b47ff [doc] Add \"mount_point_name\" request parameter in the POST share API\nd17634044 update to use openstack cli, not legacy manila cli\n9a0215101 Doc for deferred share/snapshot deletion\ndab20c306 Fixes for targeted restores in the share backup API\nb9817ebd1 Remove the confusing \u0027deferred deletion\u0027 wording from the user guide\ne6c883672 Add unit tests for CephFS driver error handling\n795ff69c5 Fix shares getting stuck in ensuring status\n3a65fee8e Replace glanceclient with openstacksdk\na6f309599 Updating release notes for share encryption feature\n52f13c16f devstack: Makes sure zfs kernel module is installed\n3039eda4a Replace eventlet.sleep with time.sleep\n65f705622 Pure Storage - bump version numbers for 2026.1 (Gazpacho)\n40107f961 Add jsonschema and bump oslo.db requirements\nd2a5271b9 Migrate setup configuration to pyproject.toml\nff3e40e6a Remove option for apache \u003c 2.4\nca9c88da0 Remove unused template file\n3b3b79214 [Doc] Annotate max API microversion in Flamingo\n79ca1c2a8 Add encryption key ref field to share create response\nf57e4307c db purge: raise on missing tables\n34c6a7335 Update master for stable/2025.2\n4de1de0f0 Stop allowing generic backups for non-NFS protocols\n754528a13 Drop eventlet usage in ssh pools\n83d088217 api: Add schema for share unmanage action\n0ae5d257a api: Add schemas for share extend, shrink actions\nb558f75b0 api: Add schemas for share soft_delete, restore actions\n25c721ca3 api: Add Schemas for share export locations\nd30785a4d Fix to create share group when only share properties are provided in filter function\necb1bbdc2 api: Combine versioned API resources\nbc7b71639 Improve scheduler performance on thin provisioning\n\nChange-Id: I2e01cb57ad5618233922d56f25e3451557f9ad2b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1c00477d2692a0bb8a9cceadcea841a1ff92383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1c00477d2692a0bb8a9cceadcea841a1ff92383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
