)]}'
{"id":"openstack%2Freleases~980212","triplet_id":"openstack%2Freleases~master~I28383bdcf72bc115c46eade02e580b12313d7497","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"13940":{"account":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"last_update":"2026-03-12 13:24:15.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-03-12 13:24:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28383bdcf72bc115c46eade02e580b12313d7497","subject":"Change Freezer to Launchpad","status":"MERGED","created":"2026-03-12 09:22:04.000000000","updated":"2026-03-12 13:24:15.000000000","submitted":"2026-03-12 13:24:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980212","meta_rev_id":"9a89b110d3315b6002be7994ee6f07195fc8667d","_number":980212,"virtual_id_number":980212,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 13:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2026-03-12 13:01:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2026-03-12 13:01:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-03-12 11:49:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 09:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 11:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2026-03-12 13:01:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"58ff564e058720fa3bd4abf3e0658b8339d6f39d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 09:22:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"effd1e604017366b8f26d0dd0484dabab633000d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:26:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"1496792bd9a4669de8f1fd39118ea5778b123dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 09:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00be426360f40ecb4d9008f03c1e558\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3382985c3da46f8a3c2361fa0b10309 : SUCCESS in 6m 21s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b381d14ad42046a19c7db27d97434015 : SUCCESS in 6m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b856d891824549f188921c6e87fc8325 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8235e276ede78b922c259a2241e16044c5dcdca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 11:45:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8f7d5f0a2cf820ab85b59424deabb58e571cde8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-12 11:47:59.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"42c6a6ca1d812490a93a81947aca6712e6fad446","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 11:49:25.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"b6b0fe67b4bc79c8d0d13aa5341a8688c7e36ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:13:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c59da03e894a4ff0be00006a982aac4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24ede3af7164dc98d887da4a56f44c3 : SUCCESS in 7m 14s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/cdd62c7ff351455a8b0b9f6166b0fcfe : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2da4b6ad339846c29f063f30df9155b1 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a47702bde998808be8c4415bc116a4a897f57a64","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-03-12 13:01:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"08ee6695eb00f32ca6694f1a09baf64979bc6a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:02:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b73451e65d5a4de50fb4abcd2f5bd43ba82ec2d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:24:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d5dac6ba0947e793913528ffb4907d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dca4392692847c2900e6efd7fa37fdc : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6ea7d5f7964801b52fda9f0cd813b4 : SUCCESS in 7m 10s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3c06feff0a65415d9863f16c6c355068 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9a89b110d3315b6002be7994ee6f07195fc8667d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:24:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"94681c84dbe22f27cab0b96227fe77bf2342b543","revisions":{"6d58ed2728ff16499a9f3a4f1d0877e438d232e6":{"kind":"REWORK","_number":1,"created":"2026-03-12 09:22:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/980212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/980212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/980212/1"}}},"commit":{"parents":[{"commit":"51b9926f8bead65d35e506ff243fc26af5b6820f","subject":"Merge \"Release neutron-vpnaas-dashboard RC1 for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51b9926f8bead65d35e506ff243fc26af5b6820f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 09:21:57.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 09:21:57.000000000","tz":60},"subject":"Change Freezer to Launchpad","message":"Change Freezer to Launchpad\n\nFreezer has switched from storyuboard to launchpad [1]\n\n[1] https://review.opendev.org/c/openstack/project-config/+/941138/\n\nChange-Id: I28383bdcf72bc115c46eade02e580b12313d7497\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d58ed2728ff16499a9f3a4f1d0877e438d232e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d58ed2728ff16499a9f3a4f1d0877e438d232e6"}]},"branch":"refs/heads/master"},"eab194a6f631d8bfc5ab1ac27ffe7772002f8a9a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 11:45:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/980212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/980212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/980212/2"}}},"commit":{"parents":[{"commit":"d834baca911ecf6b9a2f0809d2290b60acfd8a67","subject":"Merge \"Release freezer RC1 for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d834baca911ecf6b9a2f0809d2290b60acfd8a67"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 09:21:57.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 11:45:38.000000000","tz":0},"subject":"Change Freezer to Launchpad","message":"Change Freezer to Launchpad\n\nFreezer has switched from storyuboard to launchpad [1]\n\n[1] https://review.opendev.org/c/openstack/project-config/+/941138/\n\nChange-Id: I28383bdcf72bc115c46eade02e580b12313d7497\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eab194a6f631d8bfc5ab1ac27ffe7772002f8a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eab194a6f631d8bfc5ab1ac27ffe7772002f8a9a"}]},"branch":"refs/heads/master","description":"Rebase"},"94681c84dbe22f27cab0b96227fe77bf2342b543":{"kind":"NO_CHANGE","_number":3,"created":"2026-03-12 11:47:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/980212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/980212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/980212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/980212/3"}}},"commit":{"parents":[{"commit":"d834baca911ecf6b9a2f0809d2290b60acfd8a67","subject":"Merge \"Release freezer RC1 for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d834baca911ecf6b9a2f0809d2290b60acfd8a67"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 09:21:57.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-12 11:46:36.000000000","tz":60},"subject":"Change Freezer to Launchpad","message":"Change Freezer to Launchpad\n\nFreezer has switched from storyuboard to launchpad [1]\n\n[1] https://review.opendev.org/c/openstack/project-config/+/941138/\n\nChange-Id: I28383bdcf72bc115c46eade02e580b12313d7497\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94681c84dbe22f27cab0b96227fe77bf2342b543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94681c84dbe22f27cab0b96227fe77bf2342b543"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
