)]}'
{"id":"openstack%2Freleases~982898","triplet_id":"openstack%2Freleases~master~Iac9079b1104ec24aa27db2519e21a10bbfb61dbb","project":"openstack/releases","branch":"master","topic":"gazpacho-final","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-01 14:35:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac9079b1104ec24aa27db2519e21a10bbfb61dbb","subject":"Mark 2026.1 Gazpacho as released","status":"MERGED","created":"2026-03-31 18:26:36.000000000","updated":"2026-04-01 14:35:16.000000000","submitted":"2026-04-01 14:35:16.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982898-gazpacho-final","meta_rev_id":"6398684eca8eff5374d4fa5b916288350f6b0b79","_number":982898,"virtual_id_number":982898,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2026-04-01 14:35:16.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-01 11:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-01 12:43:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2026-04-01 14:35:16.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":1,"date":"2026-03-31 19:45:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-01 12:43:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2026-04-01 14:35:16.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-01 11:41:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 18:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 19:45:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-04-01 11:41:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-01 12:43:22.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2026-04-01 14:33:15.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2026-04-01 14:35:16.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"26efb1eec6aab89a308688642ffff14f11b945c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-03-31 18:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec832dc405d84ecc2cb6375fdf3d17a33144fe02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 18:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2689f9c4d44847798d5ff152c6ae69bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd51050286ee4014801e37ef260c5b74 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a91832273e420eaf9c105b3422fe24 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4065ba6e6434967bb6afeb4efca70f4 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1349b5882914565ac6d6a42ac997767 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/124892e7782045c3b10e4a963b86e7a6 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0efb816311c4a57a027020ecb331506 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e9bc662c37e49a2b712be01d049ed14 : SUCCESS in 6m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a61209220864fc9899f8424d5d54f83 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeed522511074173a58e6102482c7fb3 : SUCCESS in 7m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/dc5ffaaad052487b96ef5ceaaeffe91d : SUCCESS in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e8ae13e0d26949cea20d5acedab86a49 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"17998f224850183134633e96f40cf227bf453114","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-31 19:45:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"859c73dd8dd75a2311e843ce047d1c9e49ecce7a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 11:41:00.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e448c05490b8281985250d725641c7f78ab58440","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-04-01 12:43:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29324f3386f70515e74f49b045ce2f4b57dab746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 12:43:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9f144240e7eb1a49674c8e261972a2638623488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 12:59:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa6cbcfb535c4e22b228bdb7a1c74935\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ce05cd6f1d46c89db8ec63a6369888 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f89d376016942ff928a573a18204630 : FAILURE in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d796c610a5a64f8a8f88c2987cdddaec : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0cc411c15d04bb4933ca0dcb7917ca7 : FAILURE in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc47e80f18ce47f2807e645cf56aee14 : FAILURE in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5996d05f1d7e4d06b5842d5f3932aebc : FAILURE in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d610920a0f4695a45a173205407db9 : FAILURE in 2m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fba8b017bceb4216aa469ab2d90fddb0 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"38cc7997eafdb11af9ae466e4584a9cc0891b455","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-04-01 13:01:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f1a1eb746c2ec05ec827e45d8ea0ee1181a9772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66dfda2687374f0bb1e8fdbfcf37df58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7441c1e87aa742bb94f59524b9f48e8e : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d92ffef938e4a4fa6b11dd938fcdfda : FAILURE in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/085e1a8a431b42e0a5568f10e21ba2ab : FAILURE in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/577d041f5c9a49ffbc936fbe0449bf75 : FAILURE in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38b7034e0b864f528b147745fbe1a26b : FAILURE in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/271d0a2876874e6eb82721183b6dbfea : FAILURE in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9d2e6e637724aaaa1f96606feec839e : FAILURE in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e40ade5c6c79416d90ee1aed71877985 : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9e8ab44fde45e3bc776ef6f4d27e67 : FAILURE in 5m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2b18e4bd775948689cfd3e161ee31c71 : SUCCESS in 3m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bd04ac2ee2ba4fddae51a2849f66e956 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6d90b9a745de3aa39a854ca6c9a29900e97c2828","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-01 14:23:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cf186032cf9dc3b4c925d4e189add3b0861cdf6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2026-04-01 14:33:15.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_22348\u003e with the following votes:\n\n* Verified-1 by \u003cGERRIT_ACCOUNT_22348\u003e\n","accounts_in_message":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"_revision_number":1},{"id":"b3edf5bfb279746556e73f0d12b09a50c849ec43","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2026-04-01 14:35:16.000000000","message":"Patch Set 1: Code-Review+2 Verified+2 Workflow+1\n\nBypassing testing to merge for now, as this passed in check and the failing operations are due to a known outage, while release checks have already been manually performed.","accounts_in_message":[],"_revision_number":1},{"id":"6398684eca8eff5374d4fa5b916288350f6b0b79","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2026-04-01 14:35:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9940d37460a8b2e1bafbf3e8ce0ca3855c777c6a","revisions":{"9940d37460a8b2e1bafbf3e8ce0ca3855c777c6a":{"kind":"REWORK","_number":1,"created":"2026-03-31 18:26:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/982898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/982898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/982898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/982898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/982898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/982898/1"}}},"commit":{"parents":[{"commit":"d371eb434f6d460d6239108ea119329d3d49168c","subject":"Add release note links for 2026.1 Gazpacho #2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d371eb434f6d460d6239108ea119329d3d49168c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-31 18:25:30.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-03-31 18:26:33.000000000","tz":120},"subject":"Mark 2026.1 Gazpacho as released","message":"Mark 2026.1 Gazpacho as released\n\nAnd set 2026.2 Hibiscus as the default cycle under development.\n\nChange-Id: Iac9079b1104ec24aa27db2519e21a10bbfb61dbb\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9940d37460a8b2e1bafbf3e8ce0ca3855c777c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9940d37460a8b2e1bafbf3e8ce0ca3855c777c6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
