)]}'
{"id":"openstack%2Freleases~983055","triplet_id":"openstack%2Freleases~master~Ibf0be89a6f86d64e384cb15465e20681d2cd6506","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-01 14:38:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibf0be89a6f86d64e384cb15465e20681d2cd6506","subject":"[CI] Temporarily reduce zuul jobs number","status":"ABANDONED","created":"2026-04-01 13:16:34.000000000","updated":"2026-04-01 14:38:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd972dc5a340444957fcfb3eba8d5e007373d5d4","_number":983055,"virtual_id_number":983055,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-01 13:56:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 13:16:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-01 13:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cb958be7f40f4efee8accce1ec6b5c4409dbe9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 13:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ebe38fa76df8eac165b495359132862e17a7b6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:16:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2572ad9f82feb835d84830331e145ddf638a29e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:17:14.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/releases on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027gate\u0027][\u0027jobs\u0027]\n\nThe problem appears in the following project stanza:\n\n  project:\n      templates:\n        #- openstack-python3-jobs\n        - build-openstack-docs-pti\n      check:\n        jobs:\n          #- releases-tox-validate\n          #- releases-tox-list-changes\n          #- openstack-tox-py38:\n  ...\n\n  in \"openstack/releases/.zuul.yaml@master\", line 20","accounts_in_message":[],"_revision_number":1},{"id":"33e009e252f894877b8ce756ba321e3da2d2ef7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 13:21:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cfaa7a73b48e5ff6430e55f1c99154812ee03ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:35:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a578ab44bab42fda20c5d3de5594d02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9425ce953f54432aab4ffa2493373ee : FAILURE in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d0f40d2c214d869ff7d57e68428939 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b78fd5225e01fb4bc7811883219edac5a8e4305b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 13:46:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"09b8ea17a903f3c677792f904628f89502b38841","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:47:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4aba98fd53b2570998ae8f0c9ebab352168e4aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:48:45.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/releases on branch master.  The\nproblem was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe problem appears in the following project stanza:\n\n  project:\n      templates:\n        #- openstack-python3-jobs\n        #- build-openstack-docs-pti\n      check:\n        jobs:\n          #- releases-tox-validate\n          #- releases-tox-list-changes\n          #- openstack-tox-py38:\n  ...\n\n  in \"openstack/releases/.zuul.yaml@master\", line 20","accounts_in_message":[],"_revision_number":3},{"id":"e79a830b036cd65d3afbe8ae8c502a63f3238858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 13:51:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ae2a27bbf346d289f0c228b800aff1eef5e2611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:56:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89470dc61bd24299a53cdb4f459bd038\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6e842b2d3344498d0c2de9ce340af0 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":4},{"id":"cd972dc5a340444957fcfb3eba8d5e007373d5d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 14:38:50.000000000","message":"Abandoned\n\nno need for this as with admin rights the patch was able to merge.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3a291ed34619b296a987184581063a5aa41b3777","revisions":{"e07059dfa2e341f4aa3c893dd8feee8be437b00c":{"kind":"REWORK","_number":1,"created":"2026-04-01 13:16:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/983055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/983055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/983055/1"}}},"commit":{"parents":[{"commit":"8a301413794612fc7b198d73f107a52a6d089282","subject":"Merge \"2026.1 Gazpacho final releases for cycle-with-rc projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a301413794612fc7b198d73f107a52a6d089282"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:13:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:16:24.000000000","tz":120},"subject":"[CI] Temporarily reduce zuul jobs number","message":"[CI] Temporarily reduce zuul jobs number\n\nTo be able to land 2026.1 Gazpacho patch [1].\n\n[1] Iac9079b1104ec24aa27db2519e21a10bbfb61dbb\n\nChange-Id: Ibf0be89a6f86d64e384cb15465e20681d2cd6506\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e07059dfa2e341f4aa3c893dd8feee8be437b00c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e07059dfa2e341f4aa3c893dd8feee8be437b00c"}]},"branch":"refs/heads/master"},"c4630770ba49d4a1276922b8d418573767a688f5":{"kind":"REWORK","_number":2,"created":"2026-04-01 13:21:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/983055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/983055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/983055/2"}}},"commit":{"parents":[{"commit":"8a301413794612fc7b198d73f107a52a6d089282","subject":"Merge \"2026.1 Gazpacho final releases for cycle-with-rc projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a301413794612fc7b198d73f107a52a6d089282"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:13:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:21:09.000000000","tz":120},"subject":"[CI] Temporarily reduce zuul jobs number","message":"[CI] Temporarily reduce zuul jobs number\n\nTo be able to land 2026.1 Gazpacho patch [1].\n\n[1] Iac9079b1104ec24aa27db2519e21a10bbfb61dbb\n\nChange-Id: Ibf0be89a6f86d64e384cb15465e20681d2cd6506\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4630770ba49d4a1276922b8d418573767a688f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c4630770ba49d4a1276922b8d418573767a688f5"}]},"branch":"refs/heads/master"},"14d2aacffdf2d4445d1af1f23926e9baf85342ec":{"kind":"REWORK","_number":3,"created":"2026-04-01 13:46:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/983055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/983055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/983055/3"}}},"commit":{"parents":[{"commit":"8a301413794612fc7b198d73f107a52a6d089282","subject":"Merge \"2026.1 Gazpacho final releases for cycle-with-rc projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a301413794612fc7b198d73f107a52a6d089282"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:13:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:45:59.000000000","tz":120},"subject":"[CI] Temporarily reduce zuul jobs number","message":"[CI] Temporarily reduce zuul jobs number\n\nTo be able to land 2026.1 Gazpacho patch [1].\n\n[1] Iac9079b1104ec24aa27db2519e21a10bbfb61dbb\n\nChange-Id: Ibf0be89a6f86d64e384cb15465e20681d2cd6506\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d2aacffdf2d4445d1af1f23926e9baf85342ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14d2aacffdf2d4445d1af1f23926e9baf85342ec"}]},"branch":"refs/heads/master"},"3a291ed34619b296a987184581063a5aa41b3777":{"kind":"REWORK","_number":4,"created":"2026-04-01 13:51:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/983055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/983055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/983055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/983055/4"}}},"commit":{"parents":[{"commit":"8a301413794612fc7b198d73f107a52a6d089282","subject":"Merge \"2026.1 Gazpacho final releases for cycle-with-rc projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a301413794612fc7b198d73f107a52a6d089282"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:13:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 13:51:08.000000000","tz":120},"subject":"[CI] Temporarily reduce zuul jobs number","message":"[CI] Temporarily reduce zuul jobs number\n\nTo be able to land 2026.1 Gazpacho patch [1].\n\n[1] Iac9079b1104ec24aa27db2519e21a10bbfb61dbb\n\nChange-Id: Ibf0be89a6f86d64e384cb15465e20681d2cd6506\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a291ed34619b296a987184581063a5aa41b3777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a291ed34619b296a987184581063a5aa41b3777"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
