)]}'
{"id":"openstack%2Freleases~983074","triplet_id":"openstack%2Freleases~master~I01e2d83ec44ddbb96225493f4e68a694630547f7","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-21 09:54:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I01e2d83ec44ddbb96225493f4e68a694630547f7","subject":"DNM: release-test with required-repo","status":"ABANDONED","created":"2026-04-01 14:36:28.000000000","updated":"2026-04-21 09:54:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4514a9bf922051743dcfd43f2001d040f6d355db","_number":983074,"virtual_id_number":983074,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-01 14:49:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 14:49:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ffac8cda2990f2c53c0dd38b44b5c887f27498f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-01 14:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f682e09f05d4874c0125c368db7cd42543595a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:49:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf339f0e3f9e4f0dbeff691702e27d21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c787f1a02dbe4a37af9a8d899af2e4f8 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d100c71cea44b25913b196b7254b06f : FAILURE in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3451a0d71e047c78aebae1ea3089d58 : FAILURE in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06242b01084a410c9725cb612658532d : FAILURE in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc2d49c3c20340b9b98f5a36bbc20ff4 : FAILURE in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47ce7ef2a3a149e1b22303a019213fe5 : FAILURE in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8514dead1d8c4f3b82f3da6c87a91a9c : FAILURE in 5m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ffab69c198774a8da2114b2c3955bf37 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ccde0999fad4010a487f6e35211a35a : FAILURE in 4m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/27464831a63b44b2bcf6e50164d329a6 : SUCCESS in 3m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/62180a148d824932b59c5cb166b95f5b : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4514a9bf922051743dcfd43f2001d040f6d355db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-21 09:54:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6337759d89bfc9b9e397f1c9cf829f555995cc8","revisions":{"b6337759d89bfc9b9e397f1c9cf829f555995cc8":{"kind":"REWORK","_number":1,"created":"2026-04-01 14:36:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/983074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/983074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/983074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/983074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/983074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/983074/1"}}},"commit":{"parents":[{"commit":"8a301413794612fc7b198d73f107a52a6d089282","subject":"Merge \"2026.1 Gazpacho final releases for cycle-with-rc projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a301413794612fc7b198d73f107a52a6d089282"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 14:34:54.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-01 14:36:25.000000000","tz":120},"subject":"DNM: release-test with required-repo","message":"DNM: release-test with required-repo\n\nChange-Id: I01e2d83ec44ddbb96225493f4e68a694630547f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6337759d89bfc9b9e397f1c9cf829f555995cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6337759d89bfc9b9e397f1c9cf829f555995cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
