)]}'
{"id":"openstack%2Freleases~984971","triplet_id":"openstack%2Freleases~master~I4f4ea5e5663afd5de145406b1609ae770291182b","project":"openstack/releases","branch":"master","topic":"dalmatian-eol","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-20 10:24:22.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2026-04-30 11:48:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f4ea5e5663afd5de145406b1609ae770291182b","subject":"[skyline] Transition 2024.2 Dalmatian to End of Life","status":"MERGED","created":"2026-04-16 15:47:36.000000000","updated":"2026-04-30 11:48:43.000000000","submitted":"2026-04-30 11:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984971-dalmatian-eol","meta_rev_id":"c6c9f3eae28934d60157c75e4157c939b53086a7","_number":984971,"virtual_id_number":984971,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 11:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-04-30 11:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-30 11:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-30 11:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 15:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-16 15:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-16 16:21:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"}],"messages":[{"id":"38ffb1632d2b29269b087a5177275550b2b0c982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-16 15:47:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a23ffa3c928d68a79029b025545e959221ef5269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 15:58:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/868e63f7861c4893a6a34c719890f443\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a5bffdb4e2480a98ee2fc656172137 : SUCCESS in 3m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ea3351faaf7e44df98817a52b6e19a93 : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3a04e27189ed402f9f6396c52a63c4bd : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"32698626c7eb1bc64a19d9cd697816d7c06c4d68","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2026-04-17 02:07:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2fb2affef3f397bdef27c642e66296b32678f3f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 02:39:09.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d823670e545552584696b20a3a34200b7eb1d274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-20 09:47:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4515b3f9e6cf1b3f60221f92bdc9b1947a7b5528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:02:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41eb80a2bcea48cca17159e7425a148d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c215a0d0af7c45d28fe8fe207858e901 : SUCCESS in 7m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4f7416e5dc2041c6a4e6d03c228becac : SUCCESS in 5m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c34d194214f44d6ba8d45a67cb622e7c : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3823c42df1dd2e11d1b19ab390fc30217fe0672b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-20 10:24:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecbad6f50cb3dec879166d684bbcb0b15a35e97b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 11:37:30.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bf8986003350a26dfff31a0de7d36d5af42f8db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 11:37:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1a8fb7ff81c202f55ac9e17a176b5d2f1768adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 11:48:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d2867d95cdc4ce3b46371d801d209ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ed1b984df249c1a9d079e9467e1efe : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a4ec3f65c54888b4b8aa126629c903 : SUCCESS in 6m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1db89496222a4a9f8c0926b02f545063 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c6c9f3eae28934d60157c75e4157c939b53086a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 11:48:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99e19f203ad597e238dc32cba9db7cf623d1f499","revisions":{"3dfc2913520039e2ff58466e9ab1a5c2a59eeae0":{"kind":"REWORK","_number":1,"created":"2026-04-16 15:47:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/984971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/984971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/984971/1"}}},"commit":{"parents":[{"commit":"121615e5136cc3b69627c3b0e257bb5c7d7aa043","subject":"Merge \"Release OpenStack-Ansible 2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/121615e5136cc3b69627c3b0e257bb5c7d7aa043"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:05.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:05.000000000","tz":120},"subject":"[skyline] Transition 2024.2 Dalmatian to End of Life","message":"[skyline] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I4f4ea5e5663afd5de145406b1609ae770291182b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3dfc2913520039e2ff58466e9ab1a5c2a59eeae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3dfc2913520039e2ff58466e9ab1a5c2a59eeae0"}]},"branch":"refs/heads/master"},"99e19f203ad597e238dc32cba9db7cf623d1f499":{"kind":"REWORK","_number":2,"created":"2026-04-20 09:47:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/984971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/71/984971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/71/984971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/71/984971/2"}}},"commit":{"parents":[{"commit":"96945329d101792b22648c369df212a918decade","subject":"Merge \"Release skyline-console 5.0.1, 6.0.1, 7.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96945329d101792b22648c369df212a918decade"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:05.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-20 09:47:30.000000000","tz":120},"subject":"[skyline] Transition 2024.2 Dalmatian to End of Life","message":"[skyline] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I4f4ea5e5663afd5de145406b1609ae770291182b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99e19f203ad597e238dc32cba9db7cf623d1f499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/99e19f203ad597e238dc32cba9db7cf623d1f499"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
