)]}'
{"id":"openstack%2Freleases~984974","triplet_id":"openstack%2Freleases~master~I8f38d1e58cdba200a6226d84afa6b12adb9397de","project":"openstack/releases","branch":"master","topic":"dalmatian-eol-patch1","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2026-04-30 22:11:06.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-30 22:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-04-30 22:11:06.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2026-04-30 22:11:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f38d1e58cdba200a6226d84afa6b12adb9397de","subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","status":"MERGED","created":"2026-04-16 15:48:20.000000000","updated":"2026-04-30 22:11:06.000000000","submitted":"2026-04-30 22:11:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984974-dalmatian-eol-patch1","meta_rev_id":"3ba178c3fddeb82cef733f0458f6a62b87767f69","_number":984974,"virtual_id_number":984974,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 22:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-04-30 22:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-30 22:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-30 22:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 15:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-16 16:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-16 16:21:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2026-04-16 16:21:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-04-29 18:32:43.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"}],"messages":[{"id":"ac9efc5671a83077cc0d8cf4e1b003102dbf8886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-16 15:48:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a69682432a2bd3ef9bd0b914b4601668807c4d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:01:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac0eae218cf4c8cbc25a1c9cb1c5da1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d223f3c0fa864320aa36bf97631be36e : SUCCESS in 7m 01s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c96f4cff6b3b4c0fa747bd1f641bd64e : SUCCESS in 4m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d652beaf8af44c78b58afab28e18c582 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"72e75c7056ae9a1debed47566341279542d47d5e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-22 15:19:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fea55409e771de084678c35f21eed1300ef2969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-23 13:17:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7496661ee44516c9ec4f5e3f47bd67fcfe53c60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-23 13:18:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9e03adf47fb114b7e35f46b16ec0271e9c3b2bac","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-23 13:19:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb3cd40bcb9d14039799f5ce2a24dab90adc3c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-23 13:19:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"41c529522d8307f7f47e906fd5df96f956c1f960","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-23 13:20:05.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80bc99374c0668e7a4d8ec51ce26a691abf548b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:33:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ee0646d07a44407834fdaa2d7c89338\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b51234b9f8b4574bf7bd3555e1afce0 : SUCCESS in 8m 04s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/708dc4bfb39742c79a4aa5224d38c834 : FAILURE in 9m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e7ac7488851e486794f44dfd7988fbb1 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"36ec76a048fb14b08acfc3b16872482ca9d45c61","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-24 07:58:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c664a3c8190500a859e1b36de6d91c82574658bd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-24 10:24:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07527046429357e4231b0e3577719e3cd11c45e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-27 06:36:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3e47087ba981df58d84a9622ea18a0e7f1e418c4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-27 06:37:13.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"087ab388cbe865a3a7b8ccebfa90eb2bca88418e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 06:45:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7b8613d2c148c2a0652b2bfab7a37e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01efab4cea04032b26b63d687c622b4 : SUCCESS in 7m 56s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6dc3fbfa75284d4b847ce4e094e18076 : SUCCESS in 3m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a4e1a05957d845799673801a00927e7d : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":5},{"id":"9095aa52d44769401db2ef5b5c66af211d9b906f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-04-29 18:32:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0e794ea046b08dda91851ecdc10deda0553f390e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 18:43:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcca2d5f55cb461e86051c761a1a8828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1de7fc1ddcb4dccb2589dcd18898f28 : SUCCESS in 7m 45s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/35fa5998b90c415f8fa554c81e4e3856 : SUCCESS in 7m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/63850ea82e714ce08f0c1c9efb4c8ca8 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":6},{"id":"a3843d1903381391a7bbd4bacc6811acd79f254d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-04-30 07:27:53.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"aa2826d5e65dd399d591c3504d2845fc9a1f9ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 16:10:46.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"567bc223f8f2f27823d168b2ccbb604db1c18039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:21:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/485ba3d9f2c0461388701ca3e0146c57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866a3dcaaf444bb68df3eafcfbf7ff4b : SUCCESS in 6m 27s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/872c4994934143ffa821b7a2de16d843 : FAILURE in 7m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/220a7cbf9a0c4ee592da24895d115571 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":7},{"id":"e959b82bbd99b94cef1cca2275e00aec50b8972f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 16:27:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d8b2c36db82db13c3b896f8371938e3bac3e2ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:38:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807eed34ea4942a5aac0f7fa1a895fa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b70a1c483a4a5386b6f9b97c92ae56 : SUCCESS in 5m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3f53d25035e64cf3aa7b99d927f062ce : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/847edacceb2b455b854a27dacd98b217 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":8},{"id":"7e79c707290eea40a7d9aa15edb2027bd74b64c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-04-30 18:04:21.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"4ea33bc7616e25bb0f1dea72d607cb131f945a58","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-04-30 18:06:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3eb8da24ad5a594575454bb907ba8f3d2f69451e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 18:12:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a8fc1ed9c8484f8be9244571275396\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ef09a9dc3949d599d1f9078f6bab29 : SUCCESS in 5m 07s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7cc634ac2f0f4e00b0a69469da8aa834 : SUCCESS in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6a18c4ca96ac44328c2dd131446662df : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":9},{"id":"c0ca2efbb48d9ee0e904c9d43ea811532f857e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-04-30 21:38:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"c8ea93582d6707af7e3c2b85ebcd48036ea311b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 21:51:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/269df9c007e94abd800c9380e498a36e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718bdad5ac9c4275afbf03fc035b7fe4 : SUCCESS in 7m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/522d851cf7c146baae42915e94ef5655 : SUCCESS in 7m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fe212719d4224f52b9f8ede0256a82cd : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":10},{"id":"5f279320953c9f6d5ae4144166f0adc1bfd6f8d9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 22:00:15.000000000","message":"Patch Set 10: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7b945caf54993844743a0cc5ddcf9100209bc258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 22:00:24.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0ba4ab51e4b717286099c95691b6813c6849d998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 22:11:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfbad5999754e6cbbfdacede8882cc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70795e3543e9458cad4558ed567b9d8f : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9b6dea0571401bb41c8a60caebfeed : SUCCESS in 7m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5fc88cc221804eeeb54d4828dafa195e : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":10},{"id":"3ba178c3fddeb82cef733f0458f6a62b87767f69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 22:11:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b0850bc53add9fa70498b4f500d4311b5db57618","revisions":{"263925201026a456992e6541a61638871b42ef7d":{"kind":"REWORK","_number":1,"created":"2026-04-16 15:48:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/984974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/1"}}},"commit":{"parents":[{"commit":"121615e5136cc3b69627c3b0e257bb5c7d7aa043","subject":"Merge \"Release OpenStack-Ansible 2025.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/121615e5136cc3b69627c3b0e257bb5c7d7aa043"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/263925201026a456992e6541a61638871b42ef7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/263925201026a456992e6541a61638871b42ef7d"}]},"branch":"refs/heads/master"},"dde3871257b2b749d60214f07a043b9b73797d15":{"kind":"REWORK","_number":2,"created":"2026-04-23 13:17:23.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/74/984974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/2"}}},"commit":{"parents":[{"commit":"8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3","subject":"Merge \"nova: Release 2024.2 Dalmatian 30.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-04-23 13:17:18.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dde3871257b2b749d60214f07a043b9b73797d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dde3871257b2b749d60214f07a043b9b73797d15"}]},"branch":"refs/heads/master"},"93635a5e995a55becdac24dd3b6bc918b066f647":{"kind":"REWORK","_number":3,"created":"2026-04-23 13:18:16.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/74/984974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/3"}}},"commit":{"parents":[{"commit":"8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3","subject":"Merge \"nova: Release 2024.2 Dalmatian 30.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-04-23 13:18:10.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93635a5e995a55becdac24dd3b6bc918b066f647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93635a5e995a55becdac24dd3b6bc918b066f647"}]},"branch":"refs/heads/master"},"5004caa3ac7d3ade9b962ceb2952092270c51dce":{"kind":"REWORK","_number":4,"created":"2026-04-23 13:19:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/74/984974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/4"}}},"commit":{"parents":[{"commit":"8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3","subject":"Merge \"nova: Release 2024.2 Dalmatian 30.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8b0d13a9e4fb1a990fdb50c09a32576d47b3a4b3"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-04-23 13:19:45.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5004caa3ac7d3ade9b962ceb2952092270c51dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5004caa3ac7d3ade9b962ceb2952092270c51dce"}]},"branch":"refs/heads/master"},"2638835031f2acdd0c1387debd7b51c8d7592619":{"kind":"REWORK","_number":5,"created":"2026-04-27 06:36:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/74/984974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/5"}}},"commit":{"parents":[{"commit":"a7edef9824a9246729f3355606396c8d956843c1","subject":"Ironic final dalmatian releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7edef9824a9246729f3355606396c8d956843c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-04-27 06:36:14.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2638835031f2acdd0c1387debd7b51c8d7592619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2638835031f2acdd0c1387debd7b51c8d7592619"}]},"branch":"refs/heads/master"},"ba2ea470dd33d71af75e7f8a02a54de65ef20eae":{"kind":"REWORK","_number":6,"created":"2026-04-29 18:32:43.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/74/984974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/6"}}},"commit":{"parents":[{"commit":"a7edef9824a9246729f3355606396c8d956843c1","subject":"Ironic final dalmatian releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7edef9824a9246729f3355606396c8d956843c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-04-29 18:32:23.000000000","tz":60},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba2ea470dd33d71af75e7f8a02a54de65ef20eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba2ea470dd33d71af75e7f8a02a54de65ef20eae"}]},"branch":"refs/heads/master"},"f64aa099dc58458c6fc14ce1c5dd2c6c4170d94d":{"kind":"REWORK","_number":7,"created":"2026-04-30 16:10:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/984974/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/7"}}},"commit":{"parents":[{"commit":"a7edef9824a9246729f3355606396c8d956843c1","subject":"Ironic final dalmatian releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7edef9824a9246729f3355606396c8d956843c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-30 16:10:34.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f64aa099dc58458c6fc14ce1c5dd2c6c4170d94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f64aa099dc58458c6fc14ce1c5dd2c6c4170d94d"}]},"branch":"refs/heads/master"},"85761653ed00110df2e57530cd575487f719f29f":{"kind":"REWORK","_number":8,"created":"2026-04-30 16:27:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/984974/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/8"}}},"commit":{"parents":[{"commit":"a7edef9824a9246729f3355606396c8d956843c1","subject":"Ironic final dalmatian releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7edef9824a9246729f3355606396c8d956843c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-30 16:26:45.000000000","tz":120},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/85761653ed00110df2e57530cd575487f719f29f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/85761653ed00110df2e57530cd575487f719f29f"}]},"branch":"refs/heads/master"},"4c4f26a01ca3a0f85b19c6933339a30ae468fad8":{"kind":"REWORK","_number":9,"created":"2026-04-30 18:04:21.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/74/984974/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/9"}}},"commit":{"parents":[{"commit":"a7edef9824a9246729f3355606396c8d956843c1","subject":"Ironic final dalmatian releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7edef9824a9246729f3355606396c8d956843c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2026-04-30 18:04:09.000000000","tz":-180},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\nSigned-off-by: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4f26a01ca3a0f85b19c6933339a30ae468fad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4f26a01ca3a0f85b19c6933339a30ae468fad8"}]},"branch":"refs/heads/master"},"b0850bc53add9fa70498b4f500d4311b5db57618":{"kind":"REWORK","_number":10,"created":"2026-04-30 21:38:25.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/74/984974/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/984974/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/984974/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/984974/10"}}},"commit":{"parents":[{"commit":"b1d597332df3fde48abf49e96525b3c014757f18","subject":"Ironic Release Dalmatian 26.1.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1d597332df3fde48abf49e96525b3c014757f18"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-16 15:41:06.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2026-04-30 21:19:49.000000000","tz":-180},"subject":"[ironic] Transition 2024.2 Dalmatian to End of Life","message":"[ironic] Transition 2024.2 Dalmatian to End of Life\n\n2024.2 Dalmatian is a non-SLURP series [1], which means it moves\ndirectly to End of Life at the end of its maintained period.\nThis patch tags the tip of stable/2024.2 branches with 2024.2-eol\nand then these branches will be deleted.\n\nTeams had time to prepare final 2024.2 Dalmatian stable releases\nalready, but in case any of the deliverables needs another release\n(for example to fix a regression) before the transition, then please\nadd a -1 for this patch and propose a release patch. If the team is OK\nwith the transition then please do add a +1.\n\nThe deadline is (for non-trailing projects) April 29th, 2026.\n\n[1] https://releases.openstack.org\n\nChange-Id: I8f38d1e58cdba200a6226d84afa6b12adb9397de\nGenerated-By: openstack/releases:tools/bulk_review.sh\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\nSigned-off-by: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0850bc53add9fa70498b4f500d4311b5db57618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0850bc53add9fa70498b4f500d4311b5db57618"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
