)]}'
{"id":"openstack%2Freleases~986840","triplet_id":"openstack%2Freleases~master~Iead848e7eea5681d8cfdfdd31b793e7b5213a4f7","project":"openstack/releases","branch":"master","topic":"horizon-26.0.0","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-07-03 15:42:21.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-07-03 15:31:13.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Iead848e7eea5681d8cfdfdd31b793e7b5213a4f7","subject":"Release Horizon 26.0.0 for 2026.2 Hibiscus","status":"MERGED","created":"2026-04-30 17:30:40.000000000","updated":"2026-07-03 15:42:21.000000000","submitted":"2026-07-03 15:42:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986840-horizon-26.0.0","meta_rev_id":"fb1fbad9b9c532ef245a7031807f4a2415d9a909","_number":986840,"virtual_id_number":986840,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-03 15:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-07-03 15:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-07-03 15:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-07-02 20:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 17:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 12:03:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1a246793f57f870a3352f03901053cddcd2bccbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-04-30 17:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48fb8d59050f784f26eac1159c20ff8fdf2f9b7d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:31:28.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"1d5644ae4c448e04c4037136a3c685fccbab1fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad27ceca022846fa9f5a2d0bf4099b00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfd97be1496747939317aaeb10cd6ac5 : SUCCESS in 3m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b33300c2ccbe4459948d33bdba45168f : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/14fd3e0b29f147b284dcc13fef4dea04 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9716a416f5a2a9277437a9f8f8c55e923963bce6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-05 12:03:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b49c543532596d0b1e9ff8696889f1f945ab4e3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-06-17 19:12:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"72861a803fb29d404fa1a08fe05bf73672b004b0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 19:14:04.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f23ed2222423c30e26db6467ebcdc2e55411fac6","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-06-17 19:15:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"119cc0a3d5f327bdab3582d8f6d4df3a61b06b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 19:26:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ad3c23c7944d50af68dcd01ae786c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f997ae505924fed8e7179113739ff67 : SUCCESS in 7m 55s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9d961194d2274c608e9205f59b972aa7 : SUCCESS in 2m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ebdef8d3acff438c8dc5745fd3231da9 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"6b77aec76d5de3003bed58f27c7f883d6496e13f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-18 07:52:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec42a043c75c22d16405533c754209e40ca3f8b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-07-02 20:20:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5041aca588fbf6cd6e5cb4717d7f9230915338ca","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 20:20:59.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"883c19ca6f0ad6c36ed82bc1d16b84293e2ece9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 20:32:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607d141bd15e431cbc05d18412dff49c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02d7425a43e4515822d628e4d6997be : SUCCESS in 6m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2de11618325a4f31b258f06b3d47c0be : SUCCESS in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8a315c1a0f774fc58a30c0e3c1e1d596 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"2433524f12ec42ce450ad91ae48bd1c8859e34b7","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-07-03 02:55:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c33605586cfc1d60b8f2c071831fdbe19473dcfd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-03 15:31:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"128cf2fb8ef6bc824951e7ae376e17dd1b37dafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 15:31:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d74012c7813a2da4566508665a87e8fac87b4583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 15:42:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ffad148c3a4f79ba35e0a360326685\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d74a5bff7048d3bf343d619293ea35 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c99cc20610848dc83b9ebcd29ae1ae4 : SUCCESS in 8m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/25d4c93ebc544b95bee158c0cfc5ae9b : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":3},{"id":"fb1fbad9b9c532ef245a7031807f4a2415d9a909","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 15:42:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e49ac05a9f2dfe33d36e71594cb729ae9e27808f","revisions":{"1d0950fd1ddf10721bd02c1ea412e946c55599b4":{"kind":"REWORK","_number":1,"created":"2026-04-30 17:30:40.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/40/986840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/986840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/986840/1"}}},"commit":{"parents":[{"commit":"c70f9e9afb179356017a5550a95d54ca0bf0e5b0","subject":"Merge \"[cinder] Transition 2024.2 Dalmatian to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c70f9e9afb179356017a5550a95d54ca0bf0e5b0"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-04-30 17:29:28.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-04-30 17:29:34.000000000","tz":-300},"subject":"Release Horizon 26.0.0 for 2026.2 Hibiscus","message":"Release Horizon 26.0.0 for 2026.2 Hibiscus\n\nChange-Id: Iead848e7eea5681d8cfdfdd31b793e7b5213a4f7\nSigned-off-by: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d0950fd1ddf10721bd02c1ea412e946c55599b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1d0950fd1ddf10721bd02c1ea412e946c55599b4"}]},"branch":"refs/heads/master"},"fe336a05e824f667b440fce6d758dc9216c00b68":{"kind":"REWORK","_number":2,"created":"2026-06-17 19:12:51.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/40/986840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/986840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/986840/2"}}},"commit":{"parents":[{"commit":"1c7ca71cc6918935b2c352caa2c6fc8389b090a4","subject":"Merge \"openstacksdk 4.16.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c7ca71cc6918935b2c352caa2c6fc8389b090a4"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-04-30 17:29:28.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-06-17 19:12:32.000000000","tz":-300},"subject":"Release Horizon 26.0.0 for 2026.2 Hibiscus","message":"Release Horizon 26.0.0 for 2026.2 Hibiscus\n\nChange-Id: Iead848e7eea5681d8cfdfdd31b793e7b5213a4f7\nSigned-off-by: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe336a05e824f667b440fce6d758dc9216c00b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe336a05e824f667b440fce6d758dc9216c00b68"}]},"branch":"refs/heads/master"},"e49ac05a9f2dfe33d36e71594cb729ae9e27808f":{"kind":"REWORK","_number":3,"created":"2026-07-02 20:20:06.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/40/986840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/40/986840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/40/986840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/40/986840/3"}}},"commit":{"parents":[{"commit":"fb40d269a2d495a4d9e52bbb0002adb0d92ea89f","subject":"Merge \"Release stevedore for Hibiscus-2 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb40d269a2d495a4d9e52bbb0002adb0d92ea89f"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-04-30 17:29:28.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-07-02 20:19:32.000000000","tz":-300},"subject":"Release Horizon 26.0.0 for 2026.2 Hibiscus","message":"Release Horizon 26.0.0 for 2026.2 Hibiscus\n\nChange-Id: Iead848e7eea5681d8cfdfdd31b793e7b5213a4f7\nSigned-off-by: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e49ac05a9f2dfe33d36e71594cb729ae9e27808f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e49ac05a9f2dfe33d36e71594cb729ae9e27808f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
