)]}'
{"id":"openstack%2Freleases~992417","triplet_id":"openstack%2Freleases~master~Id79f5c449c2c9029cbe3f110da38a4336a28db83","project":"openstack/releases","branch":"master","topic":"pyproject-toml-releases","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-07-21 17:42:11.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-09 12:10:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Id79f5c449c2c9029cbe3f110da38a4336a28db83","subject":"Add pyproject.toml","status":"MERGED","created":"2026-06-09 09:56:16.000000000","updated":"2026-07-21 17:42:11.000000000","submitted":"2026-07-21 17:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992417-pyproject-toml-releases","meta_rev_id":"85635903f571bca2fd8b5770cad1432220b55da7","_number":992417,"virtual_id_number":992417,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 17:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2026-06-11 12:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-06-09 12:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-21 15:05:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-21 17:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2026-06-09 13:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 09:56:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-09 10:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 12:14:37.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2026-07-21 15:05:47.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"7f790cedf153b33adf223b088a397cc478494566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-09 09:56:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3db90e5125c8bd50ec5650e428ed29d9a5edc9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 10:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919f86cd4cfc45fdba411dd1c13245a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055ad57c27f14f66a5384cef1da1067e : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db57390d620641e7bca0cfbf220f9c05 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e626e886caa428085b2354ba316aebb : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8679e3880e547708f91df7c3d895625 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa2818338ff54911acb2348ad23c3c12 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d382fb5ea72a494e9eff832ad50550c8 : SUCCESS in 3m 58s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5c541cd2ea524ef7add4eec71ce29205 : SUCCESS in 2m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ba286c853b0d4666880385c0591bb682 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bf7b255f95ece24f5cf674717f09ad43cce2ef6b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-09 11:20:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"27cc48d226be538348be5dd787218766a361819b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-09 12:08:47.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f919ef0226773879066a52e375bcc791dfc4ab70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-09 12:08:52.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f935ff762cd519455e2f4f1e6662d03aeb1409a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-09 12:10:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"35b93f1167e10e69a8d6372caebbe09a098adf50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 12:18:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30938614d6034e308b916432e7274b86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d3762d89f1546818b4182f9710c74a1 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1acfc7705d9f4d7fad2524074b7cc9ca : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/680d1f52868d422280d9bfd749997e76 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d493ed896264225a0e206594a6b7eff : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da6fb1e77fc94fac85404c7d066cb98d : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f308ad78d9aa46388ef849f55728fffa : SUCCESS in 4m 26s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ef445021ae5343c6a23d6abfd9e900da : SUCCESS in 2m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/29b2403b3d58417aaf780ba1d6779ae5 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e878461fc871f918f97e893c292eaa02e92234b7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-09 13:27:59.000000000","message":"Patch Set 2: PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbcf90d41bf76291f6167b46f821435055ad95ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-09 14:23:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0a41493df47c79777248136480495bdefefe52d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-06-11 12:14:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5fc042b1ab1a18eb496236d4189e403ca5af0859","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-07-21 15:05:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nIt doesn\u0027t appear we publish this to PyPI, but it can\u0027t hurt to be consistent with our projects that do.","accounts_in_message":[],"_revision_number":2},{"id":"5834dcd983bdcbf7139f1f2c8b49d7f2c84feaff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:06:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42eb749947f70cba98652dab26b0e638503450de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:22:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c31ae9b29f6450db93ca8ef44290561\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0df70f1a6c4df6a2ed4a31bba67d50 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c67c36d7e6ad4303b8fd8d51ea437ee2 : FAILURE in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0ace91a3274db8a95b0855ce498fbc : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56d0d66d8f0e48319f686fa764df96fa : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3034383133014c16af5a4f35209d4eba : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d806551b75824f4e96ec90f908954d4f : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4f97e802d6457d96935f8a2220fef7 : SUCCESS in 6m 46s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/415330ed1e5040dcafe1ccaa35ce24fa : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1706da0c072ce8c0d2054b297d0ec749573a1e3d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-21 15:43:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"601cf5709922da325f2e7e42018862515c669b98","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-21 16:40:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a501892164fc17bc71e0ebc9e85b663a3c00102b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:58:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2256956b1e45bfbae7c307e62b5b8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e680999f784c03bf9c3553956ddad9 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46dbb9f0f7994681b66b15e75a84ade3 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee15714c8c1942e4b0058c2235113f1d : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c61faff5a9d4a4aa28ef42bd2d7bb42 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01da47bf2118490ca726d3a533876d30 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab2b6017ae248eea4cfda5e7d40381e : SUCCESS in 7m 31s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5625305626954831a7bba44a44181e39 : SUCCESS in 7m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4bda5c2f815a446baea9d4bb2cc06f94 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b2ffe7b705a920909b91d409c999c99dd191861d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:59:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c74e399308c03d643a02b396528801113fe7441","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-07-21 17:05:48.000000000","message":"Patch Set 2: -Workflow\n\nUnapproving since Jens has questions.","accounts_in_message":[],"_revision_number":2},{"id":"6fbf79c708c490912ef61ab9349b3f410c175543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:05:51.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"eb38babf4982f3364dae9fc809570e90311b8e5d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-07-21 17:06:59.000000000","message":"Patch Set 2: Workflow+1\n\nOops, questions were on the parent patch, I guess they\u0027ll need a new change to address.","accounts_in_message":[],"_revision_number":2},{"id":"c9997a9f445fa04fac7ce9cbad135984a1571c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:17:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9655ca49d3547de96f76500685bd99a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02b948150874584b1c7504bc6f5899d : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47d3a4159af047f9abb914102963907c : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a83b315f1d14b0686a635c417c13fc0 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12922ae59ed145b886af3230fd6aa7c1 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee3df19ec68746b596023538a73ea7da : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f998f454feb14cdeb8fc70a820f17f2e : SUCCESS in 4m 10s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fca9cc53aca54312afb26198f0ce0557 : SUCCESS in 2m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f380b44fb7524bfc91ee732c2c76fd25 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":2},{"id":"eb58cc3fcb9be4f20fe9ef0766f190c59d8ee2ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:17:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f29e47c684d3458d80287c1dd3efca7b2e27a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:42:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d20b2b07e04d87bcb4fb7c19c85782\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6baa53f3e4e34245aa52e38aabcf18de : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5892eb7ab58043538a08200f723e0144 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/815c04b33b674602a484e7aafb6ad5af : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b9eae71a1d0433abff86aa87a8af95c : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba1b1b6a446462db6f572b77ad424ad : SUCCESS in 7m 26s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/90936d481c2647ebac68fc10f37a378e : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce1890b5b1ad4cc7ad230f181825c811 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"85635903f571bca2fd8b5770cad1432220b55da7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:42:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8001a24eb3b9b7be58f81850be04bd08bf91ab8a","revisions":{"79cf6dfaec53a6aa97e9ad313dfe995136034ea2":{"kind":"REWORK","_number":1,"created":"2026-06-09 09:56:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/992417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/992417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/992417/1"}}},"commit":{"parents":[{"commit":"aaa987c79c4a0dc568f3364c7195d96c514be65d","subject":"Merge \"Release governance again with updated pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aaa987c79c4a0dc568f3364c7195d96c514be65d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-09 09:00:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-09 09:55:16.000000000","tz":120},"subject":"Add pyproject.toml","message":"Add pyproject.toml\n\nProjects are migrating to the newer pyproject.toml from the old,\nlegacy, setup.cfg. Let\u0027s do the same for openstack_releases.\n\nChange-Id: Id79f5c449c2c9029cbe3f110da38a4336a28db83\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79cf6dfaec53a6aa97e9ad313dfe995136034ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79cf6dfaec53a6aa97e9ad313dfe995136034ea2"}]},"branch":"refs/heads/master"},"8001a24eb3b9b7be58f81850be04bd08bf91ab8a":{"kind":"REWORK","_number":2,"created":"2026-06-09 12:08:47.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/992417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/992417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/992417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/992417/2"}}},"commit":{"parents":[{"commit":"aaa987c79c4a0dc568f3364c7195d96c514be65d","subject":"Merge \"Release governance again with updated pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aaa987c79c4a0dc568f3364c7195d96c514be65d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-09 09:00:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-09 12:06:25.000000000","tz":0},"subject":"Add pyproject.toml","message":"Add pyproject.toml\n\nProjects are migrating to the newer pyproject.toml from the old,\nlegacy, setup.cfg. Let\u0027s do the same for openstack_releases.\n\nChange-Id: Id79f5c449c2c9029cbe3f110da38a4336a28db83\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8001a24eb3b9b7be58f81850be04bd08bf91ab8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8001a24eb3b9b7be58f81850be04bd08bf91ab8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
