)]}'
{"id":"openstack%2Freleases~997556","triplet_id":"openstack%2Freleases~master~I8d1242700fd8b7aa1bd80540dbd651b3988e059d","project":"openstack/releases","branch":"master","topic":"epoxy-stable","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-26 10:49:57.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-08-26 11:04:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d1242700fd8b7aa1bd80540dbd651b3988e059d","subject":"Release octavia for stable/2025.1","status":"MERGED","created":"2026-07-16 13:21:38.000000000","updated":"2026-08-26 11:04:22.000000000","submitted":"2026-08-26 11:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"997556-epoxy-stable","meta_rev_id":"380223d9390dcbf844061f9760e534d9b7c1bb4a","_number":997556,"virtual_id_number":997556,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 11:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-08-26 10:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-08-26 09:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-08-26 10:49:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-08-26 09:20:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 13:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 13:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-16 15:39:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"6789e75c1d85d5bcc94aedddaf8be28e2d5f2e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-07-16 13:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c29674dc08c300c6bfad5e23735044956657ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:34:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d733fe9f1924b449df4b17da3348873\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d26c24240a749b78638889b04fde24d : SUCCESS in 8m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/138ea5f730ed481b85972be680d2b930 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6a88563664404141941acac91da88350 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"27d660c6a831e39a64df61320990369af1da2273","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-12 15:56:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be3ea838f9dc52d58393a31e13c1643bb8b0e9cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 06:53:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"89460e8323958e65cdf90ed2591542af0033acd9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 06:54:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5dcece485f1659cdc8ad089b108733257a99a2a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:54:26.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"6e70297171f86aef860fc063b2618aa6e4cfd074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 07:01:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/356d4599947c475aadffa16648e8dbcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eba4718a81342289d5e3108f9c1abc0 : SUCCESS in 4m 32s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7eca6c7045a049d4850a93b710f78e36 : SUCCESS in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b035c7fe5fbf488fa61bc1395f344bdb : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"90aac781e3b5d30dd0f80e7e2fed72968b2e3782","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-26 09:09:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"450530a3a370402e134ea181ec570c6dd764f026","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 09:19:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4647663d1974499636f00b970f8fdab6abb52a7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 09:20:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"21c76be03eaa055bdbab1ef452be35902c17f670","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-08-26 09:20:41.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1cfe8bf4d220b6ed29a3c689009964074396db21","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:20:57.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"8a83daa2485afbba35b2389710fcd79bc5e4e216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:26:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf2e5d30f3e64a86981f688f2e241494\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c3c68d3ea3405f979ec90c4a9daf05 : SUCCESS in 4m 16s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c16d8e9842ed431fbaa91df4594c8578 : SUCCESS in 4m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/44d4496af3d94135b4015f20a0184b2d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"9e1f494087d9b73565563bf19f8e11c5d3dde0be","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-26 10:49:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1459adbeecd6777f8a10deff107befbb874d5ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:51:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f15d203c4ecd5c906763bab4b6c182d52102897e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 11:04:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2bd9cb68cd465791443290d7aa1475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760d69be330d4ebfb9842fa9ca99ff99 : SUCCESS in 7m 58s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/df6e1ad4792b49ddb2f94de13b9750f5 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"380223d9390dcbf844061f9760e534d9b7c1bb4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 11:04:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dbc5d9f6ed346fb0217b3d5c2056a7a921ddbecf","revisions":{"827ef39396f10756d3cda101bdb1a01835a4d8c3":{"kind":"REWORK","_number":1,"created":"2026-07-16 13:21:38.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/997556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/997556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/997556/1"}}},"commit":{"parents":[{"commit":"11c1e16d0b42e0e465cd07bd9550c56c362fe7f5","subject":"Merge \"Release oslo.concurrency 7.6.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11c1e16d0b42e0e465cd07bd9550c56c362fe7f5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-16 13:21:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-16 13:21:34.000000000","tz":120},"subject":"Release octavia for stable/2025.1","message":"Release octavia for stable/2025.1\n\nAs a preparation for the transition of 2025.1 Epoxy to Unmaintained [1],\nthis release picks up new commits from octavia since the last\nstable/2025.1 release.\n\nThis patch is proposed as a convenience to ensure that the latest\nmerged stable changes are released. If the team is good with this,\nthen please respond with a +1 to let the release team know it is OK\nto proceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 16.0.1..0fff2867b\n1a04c5c40 Make [task_flow].persistence_connection secret\na43b5bbaa Add SKI and AKI extensions to amphora certificates\n3ac3a9de0 Remove failing drivers from enabled_provider_drivers\nc826891aa Fix issues related to pkg_resources module\n91179e6c8 Fixed missing mock in TestAmphoraAgentCMD.test_main\n8eb2e9cdd Fix database connection check\ne67c6b40b Don\u0027t fail if a provider driver cannot be loaded in Octavia API\nac8a7adb1 Move log offload files to /var/log/octavia\n39aa5a3d9 Update diskimage-create.sh to use noble as default\n\nChange-Id: I8d1242700fd8b7aa1bd80540dbd651b3988e059d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/827ef39396f10756d3cda101bdb1a01835a4d8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/827ef39396f10756d3cda101bdb1a01835a4d8c3"}]},"branch":"refs/heads/master"},"584eca1f37b318d35b88a26787c14af93b0c2e60":{"kind":"REWORK","_number":2,"created":"2026-08-26 06:53:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/997556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/997556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/997556/2"}}},"commit":{"parents":[{"commit":"11c1e16d0b42e0e465cd07bd9550c56c362fe7f5","subject":"Merge \"Release oslo.concurrency 7.6.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11c1e16d0b42e0e465cd07bd9550c56c362fe7f5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-16 13:21:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-26 06:53:20.000000000","tz":120},"subject":"Release octavia for stable/2025.1","message":"Release octavia for stable/2025.1\n\nAs a preparation for the transition of 2025.1 Epoxy to Unmaintained [1],\nthis release picks up new commits from octavia since the last\nstable/2025.1 release.\n\nThis patch is proposed as a convenience to ensure that the latest\nmerged stable changes are released. If the team is good with this,\nthen please respond with a +1 to let the release team know it is OK\nto proceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 16.0.1..9994bc7\ncf5006373 Fix HAProxy config with cpu-map and external-check\nd2d0dec72 Fix HAProxy config injection via L7 policy redirect URLs\n0c0fa9694 Fix HAProxy config injection via tls_ciphers field\nfa4fc311e Fix QoS policy validation to use request context\nb16147c11 Fix ZooKeeper session churn in ZookeeperTaskFlowDriver\n939342604 Remove failing centos-9-stream job\n1a04c5c40 Make [task_flow].persistence_connection secret\na43b5bbaa Add SKI and AKI extensions to amphora certificates\n3ac3a9de0 Remove failing drivers from enabled_provider_drivers\nc826891aa Fix issues related to pkg_resources module\n91179e6c8 Fixed missing mock in TestAmphoraAgentCMD.test_main\n8eb2e9cdd Fix database connection check\ne67c6b40b Don\u0027t fail if a provider driver cannot be loaded in Octavia API\nac8a7adb1 Move log offload files to /var/log/octavia\n39aa5a3d9 Update diskimage-create.sh to use noble as default\n\nChange-Id: I8d1242700fd8b7aa1bd80540dbd651b3988e059d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/584eca1f37b318d35b88a26787c14af93b0c2e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/584eca1f37b318d35b88a26787c14af93b0c2e60"}]},"branch":"refs/heads/master"},"dbc5d9f6ed346fb0217b3d5c2056a7a921ddbecf":{"kind":"REWORK","_number":3,"created":"2026-08-26 09:20:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/56/997556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/997556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/997556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/997556/3"}}},"commit":{"parents":[{"commit":"11c1e16d0b42e0e465cd07bd9550c56c362fe7f5","subject":"Merge \"Release oslo.concurrency 7.6.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11c1e16d0b42e0e465cd07bd9550c56c362fe7f5"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-07-16 13:21:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2026-08-26 09:20:03.000000000","tz":120},"subject":"Release octavia for stable/2025.1","message":"Release octavia for stable/2025.1\n\nAs a preparation for the transition of 2025.1 Epoxy to Unmaintained [1],\nthis release picks up new commits from octavia since the last\nstable/2025.1 release.\n\nThis patch is proposed as a convenience to ensure that the latest\nmerged stable changes are released. If the team is good with this,\nthen please respond with a +1 to let the release team know it is OK\nto proceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 16.0.1..9994bc7\ncf5006373 Fix HAProxy config with cpu-map and external-check\nd2d0dec72 Fix HAProxy config injection via L7 policy redirect URLs\n0c0fa9694 Fix HAProxy config injection via tls_ciphers field\nfa4fc311e Fix QoS policy validation to use request context\nb16147c11 Fix ZooKeeper session churn in ZookeeperTaskFlowDriver\n939342604 Remove failing centos-9-stream job\n1a04c5c40 Make [task_flow].persistence_connection secret\na43b5bbaa Add SKI and AKI extensions to amphora certificates\n3ac3a9de0 Remove failing drivers from enabled_provider_drivers\nc826891aa Fix issues related to pkg_resources module\n91179e6c8 Fixed missing mock in TestAmphoraAgentCMD.test_main\n8eb2e9cdd Fix database connection check\ne67c6b40b Don\u0027t fail if a provider driver cannot be loaded in Octavia API\nac8a7adb1 Move log offload files to /var/log/octavia\n39aa5a3d9 Update diskimage-create.sh to use noble as default\n\nChange-Id: I8d1242700fd8b7aa1bd80540dbd651b3988e059d\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbc5d9f6ed346fb0217b3d5c2056a7a921ddbecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbc5d9f6ed346fb0217b3d5c2056a7a921ddbecf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
