)]}'
{"id":"openstack%2Frenderspec~654654","triplet_id":"openstack%2Frenderspec~master~Iffceba059cc3bed33feb2e3a28721b789788b825","project":"openstack/renderspec","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Iffceba059cc3bed33feb2e3a28721b789788b825","subject":"Dropping the py35 testing","status":"ABANDONED","created":"2019-04-22 23:19:14.000000000","updated":"2019-07-18 02:45:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab43614dbb061b54e99e45007d91f751caf64619","_number":654654,"virtual_id_number":654654,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-04-22 23:34:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-17 20:16:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 23:34:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 20:16:17.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"}],"messages":[{"id":"5741392e157d8b8972613abda71e86b4554c916c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-22 23:19:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95a4f5b4aeee0628b2639679528035de9c9c550e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 23:34:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/54/654654/1/check/test-release-openstack/6f0f6e0/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/54/654654/1/check/openstack-tox-pep8/2371fda/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/54/654654/1/check/openstack-tox-py27/844a408/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/54/654654/1/check/openstack-tox-py36/27d5c2f/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/54/654654/1/check/openstack-tox-docs/2be0420/ : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e8608c5847fa172f686646aac92b74655e6fcf0a","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-17 20:16:17.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we abandon this in favor of https://opendev.org/openstack/renderspec/commit/f74f393218c35f5d16dbeed25e942338c7479d6c ?","accounts_in_message":[],"_revision_number":1},{"id":"74e29464defabb5165f4fb2435171547d8ebfbab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 02:45:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddefd6b530c46f121d0dba0222e241f8c390255b","revisions":{"ddefd6b530c46f121d0dba0222e241f8c390255b":{"kind":"REWORK","_number":1,"created":"2019-04-22 23:19:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/654654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/54/654654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/54/654654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/54/654654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/54/654654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/54/654654/1"}}},"commit":{"parents":[{"commit":"adcaafe9f2478fb43e980aa75a026c4421df3788","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/adcaafe9f2478fb43e980aa75a026c4421df3788"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 23:18:50.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 23:19:12.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Iffceba059cc3bed33feb2e3a28721b789788b825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/ddefd6b530c46f121d0dba0222e241f8c390255b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/ddefd6b530c46f121d0dba0222e241f8c390255b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
