)]}'
{"id":"openstack%2Frenderspec~668836","triplet_id":"openstack%2Frenderspec~master~I08573b53609d67ebe829bfbb3f82edd7fac43f20","project":"openstack/renderspec","branch":"master","topic":"py-versions-default","hashtags":[],"change_id":"I08573b53609d67ebe829bfbb3f82edd7fac43f20","subject":"Add new context variable \u0027py_versions_default\u0027","status":"ABANDONED","created":"2019-07-03 09:15:28.000000000","updated":"2019-07-03 12:15:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1bda7da1bdfb8eb0c32c9baafbc51c0782a5ea14","_number":668836,"virtual_id_number":668836,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-07-03 09:48:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 09:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba14f0d6b2ad01e22f4929bc2689d3a9f410d1ee","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 09:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab97d7f79671d8e35704d8194005d2d6655caff4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:48:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/668836/1/check/openstack-tox-pep8/1ccbe74/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/36/668836/1/check/openstack-tox-py27/34e1ac0/ : FAILURE in 15m 48s\n- openstack-tox-py35 http://logs.openstack.org/36/668836/1/check/openstack-tox-py35/b635c16/ : SUCCESS in 2m 23s\n- openstack-tox-py36 http://logs.openstack.org/36/668836/1/check/openstack-tox-py36/24116fa/ : FAILURE in 27m 27s\n- openstack-tox-docs http://logs.openstack.org/36/668836/1/check/openstack-tox-docs/7dd53f6/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bf0e372ac052d8f9d5d8879aafa903ebcee574d7","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 12:15:14.000000000","message":"Abandoned\n\nThis will not work. In the case of singlespec and Requires, we don\u0027t want to expand to py2 and py3 by default....","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c356499d0e8fce7565b0e05e8de64ed8d6fe97ba","revisions":{"c356499d0e8fce7565b0e05e8de64ed8d6fe97ba":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:15:28.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/36/668836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/36/668836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/36/668836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/36/668836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/36/668836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/36/668836/1"}}},"commit":{"parents":[{"commit":"10dba1749532330048bad511b2a982a85986915e","subject":"Fix _pymod2pkg_translate when skip_pyversion is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/10dba1749532330048bad511b2a982a85986915e"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 09:14:00.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 09:14:00.000000000","tz":120},"subject":"Add new context variable \u0027py_versions_default\u0027","message":"Add new context variable \u0027py_versions_default\u0027\n\nThe new variable \u0027py_versions_default\u0027 is useful for porting services\nto python3.\nWhen the variable is set, all py2pkg() calls that don\u0027t have\n\u0027py_versions\u0027 explicit set will use the \u0027py_versions_default\u0027 value.\nx\n\nChange-Id: I08573b53609d67ebe829bfbb3f82edd7fac43f20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/c356499d0e8fce7565b0e05e8de64ed8d6fe97ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/c356499d0e8fce7565b0e05e8de64ed8d6fe97ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
