)]}'
{"id":"openstack%2Frenderspec~668890","triplet_id":"openstack%2Frenderspec~master~I273919cf4f47bf9859d71bc216cbf96e82a2a6f7","project":"openstack/renderspec","branch":"master","hashtags":[],"change_id":"I273919cf4f47bf9859d71bc216cbf96e82a2a6f7","subject":"Add contextfunctions \u0027py2\u0027 and \u0027py3\u0027","status":"MERGED","created":"2019-07-03 12:30:58.000000000","updated":"2019-07-03 18:12:51.000000000","submitted":"2019-07-03 18:12:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668890-1562177571086-61dee02a","meta_rev_id":"cec695b44606c9697d89c65d64fedc3141b93772","_number":668890,"virtual_id_number":668890,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-03 18:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":2,"date":"2019-07-03 16:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-07-03 15:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 18:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 15:51:53.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-03 16:58:09.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-07-03 18:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08e0f5139d7abc444b64e70db4d313b715c36153","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 12:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f521ac818a00ded62195eadc3bdfd0b588599f11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/668890/1/check/openstack-tox-pep8/e4ceb7f/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/90/668890/1/check/openstack-tox-py27/cb00d8d/ : SUCCESS in 2m 41s\n- openstack-tox-py35 http://logs.openstack.org/90/668890/1/check/openstack-tox-py35/3418726/ : SUCCESS in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/90/668890/1/check/openstack-tox-py36/7262929/ : SUCCESS in 2m 44s\n- openstack-tox-docs http://logs.openstack.org/90/668890/1/check/openstack-tox-docs/d59bd8a/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"93355205b76504ddf2daf2609c8c1069bb2b819e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-03 15:51:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebb43c457f7878dcf11e0bc5ebf40b43d31a5241","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-07-03 16:58:09.000000000","message":"Patch Set 1: Code-Review+2\n\nwfm","accounts_in_message":[],"_revision_number":1},{"id":"9f70e27bcaac05e785819ad5d02bdc6c0bde85d0","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-03 18:04:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d6e666e2af61f1dbb29f29cb24db2ec360bebcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:04:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95a266932800c80d4ebfa1567239a1edb25c9dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:12:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/668890/1/gate/openstack-tox-pep8/a951b4e/ : SUCCESS in 2m 28s\n- openstack-tox-py27 http://logs.openstack.org/90/668890/1/gate/openstack-tox-py27/19a4244/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/90/668890/1/gate/openstack-tox-py35/bb84485/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/90/668890/1/gate/openstack-tox-py36/13312ec/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/90/668890/1/gate/openstack-tox-docs/cacad70/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cec695b44606c9697d89c65d64fedc3141b93772","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 18:12:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc","revisions":{"8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc":{"kind":"REWORK","_number":1,"created":"2019-07-03 12:30:58.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/90/668890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/90/668890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/90/668890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/90/668890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/90/668890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/90/668890/1"}}},"commit":{"parents":[{"commit":"10dba1749532330048bad511b2a982a85986915e","subject":"Fix _pymod2pkg_translate when skip_pyversion is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/10dba1749532330048bad511b2a982a85986915e"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 12:30:00.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-03 12:30:00.000000000","tz":120},"subject":"Add contextfunctions \u0027py2\u0027 and \u0027py3\u0027","message":"Add contextfunctions \u0027py2\u0027 and \u0027py3\u0027\n\nBoth are shortcuts to \u0027py2pkg\u0027 with specific py_versions set.\nThis is useful when defining BuildRequires that are eg. python3 only.\n\nChange-Id: I273919cf4f47bf9859d71bc216cbf96e82a2a6f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
