)]}'
{"id":"openstack%2Frenderspec~669475","triplet_id":"openstack%2Frenderspec~master~Iec17f01beed324e1481391204533eeafaf6253e3","project":"openstack/renderspec","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Iec17f01beed324e1481391204533eeafaf6253e3","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:22:33.000000000","updated":"2019-07-07 08:13:54.000000000","submitted":"2019-07-07 08:13:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669475-1562487234325-6efaa91b","meta_rev_id":"22e3dfb1f5a9e571ddfb57ec2cf67515a0407555","_number":669475,"virtual_id_number":669475,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-07-07 08:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-07-05 23:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-07 08:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-07 08:08:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 23:19:02.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-07 08:08:35.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-07-07 08:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c25ffe94494286e44b268079faf482902f27260","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3babcb6230a83bd9262dc0dd6771c21ebdc1a34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/75/669475/1/check/test-release-openstack/13398b3/ : SUCCESS in 1m 42s\n- openstack-tox-pep8 http://logs.openstack.org/75/669475/1/check/openstack-tox-pep8/67e361b/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/75/669475/1/check/openstack-tox-py27/c8fa8d2/ : SUCCESS in 2m 15s\n- openstack-tox-py36 http://logs.openstack.org/75/669475/1/check/openstack-tox-py36/bcf7825/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/75/669475/1/check/openstack-tox-py37/95942d3/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/75/669475/1/check/openstack-tox-docs/a0f430c/html/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c60301d48ed3cf886ee2139d06a488e0ae82cda1","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-05 23:19:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5257b98779a403013a7eea9edd00d9829e4f10c9","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-07 08:08:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0035a6d065f05f113b1140003b0089d921f7f98d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 08:08:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7f4adc2a3b4b5409fc8f8be5ed737b3eb13d299","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 08:13:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6239846c98b22a80a4b3bbf39ecfc79148a47b66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 08:13:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/75/669475/1/gate/test-release-openstack/ebf60fe/ : SUCCESS in 2m 07s\n- openstack-tox-pep8 http://logs.openstack.org/75/669475/1/gate/openstack-tox-pep8/02c4d7a/ : SUCCESS in 2m 28s\n- openstack-tox-py27 http://logs.openstack.org/75/669475/1/gate/openstack-tox-py27/40cd132/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/75/669475/1/gate/openstack-tox-py36/37ba5f0/ : SUCCESS in 2m 44s\n- openstack-tox-py37 http://logs.openstack.org/75/669475/1/gate/openstack-tox-py37/d98fbe1/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/75/669475/1/gate/openstack-tox-docs/07fb0d9/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74f393218c35f5d16dbeed25e942338c7479d6c","revisions":{"f74f393218c35f5d16dbeed25e942338c7479d6c":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:22:33.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/75/669475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/75/669475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/75/669475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/75/669475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/75/669475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/75/669475/1"}}},"commit":{"parents":[{"commit":"8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc","subject":"Add contextfunctions \u0027py2\u0027 and \u0027py3\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/8083f1ed3c66d9eb5f369db37f52d1b9e9c1eecc"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:57:20.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:22:33.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Iec17f01beed324e1481391204533eeafaf6253e3\nStory: #2005924\nTask: #34235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/f74f393218c35f5d16dbeed25e942338c7479d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/f74f393218c35f5d16dbeed25e942338c7479d6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
