)]}'
{"id":"openstack%2Frenderspec~707961","triplet_id":"openstack%2Frenderspec~master~I398df150ebf6e5ec9bc34344f0f95a6eb2d0244d","project":"openstack/renderspec","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I398df150ebf6e5ec9bc34344f0f95a6eb2d0244d","subject":"Set ignore_basepython_conflict in tox","status":"MERGED","created":"2020-02-15 23:24:03.000000000","updated":"2020-03-10 05:45:28.000000000","submitted":"2020-03-10 05:44:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707961-1583819059871-99049c42","meta_rev_id":"8c033a9c2b3c0cdd1a5f464cf08f15d45c61ef91","_number":707961,"virtual_id_number":707961,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 05:44:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2020-03-04 21:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-02-20 09:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-10 05:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-10 05:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 09:53:00.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-03-04 21:57:19.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-03-10 05:38:34.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-03-10 05:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22066f1ad71f98b6cc4201ec0a0c1f278e22fb9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-15 23:24:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32f8d439c37dffa1f08203b060ecb05cc1ecce2f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-15 23:24:32.000000000","message":"Topic set to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"57ae436d342d2c243a82347c25342d763b0d9d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 23:32:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d992930cb64b5f9c9a8ba4d6fcacee : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2bd69aa5eb440aaa95f9e8099d75de5 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dad04e813b60450a8f6aae79196539cf : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e2cd845abe4809bf02ba1e2fd51990 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55c3758db5341228b3168bc51c2a4b3 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"12d965fee6ef5496cce7dbc0c7526976c578b9d3","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-02-20 09:53:00.000000000","message":"Patch Set 1: Code-Review+1\n\nMakes sense","accounts_in_message":[],"_revision_number":1},{"id":"387bc8d1b4561086d9eb06f70efda4bb7b3fb50f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-04 21:57:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d385a80277c2c73b5eea63127ffcf5aff86a2dc4","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-03-10 05:38:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e17c07586fb8831b40da6f16a2e46d3734f5817a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 05:38:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdacdf9f3c9d07f85a8d6135a712d36966ca4618","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 05:44:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e42c45cb8f7b4dcafd49ab045309970db0be4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 05:44:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ee8e2cde5f4b0480abfa5607b57e3b : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96fa448fe1f34c9c9f9740dbf6db8deb : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5294e079f3aa4e498e42771fcbc65afc : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f71a7db723a4c349679bc852772ed32 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8c033a9c2b3c0cdd1a5f464cf08f15d45c61ef91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 05:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42909ca69999432996a91363cc04eace : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"809e959b144244f449f55b8178ca5bd403a0192c","revisions":{"809e959b144244f449f55b8178ca5bd403a0192c":{"kind":"REWORK","_number":1,"created":"2020-02-15 23:24:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/707961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/61/707961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/61/707961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/61/707961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/61/707961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/61/707961/1"}}},"commit":{"parents":[{"commit":"a9241367d71f2ff21f005797217c272dd19f7f7f","subject":"Merge \"Drop py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/a9241367d71f2ff21f005797217c272dd19f7f7f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-15 23:22:36.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-15 23:24:02.000000000","tz":-360},"subject":"Set ignore_basepython_conflict in tox","message":"Set ignore_basepython_conflict in tox\n\nIda03f9cca7d276ed8a449d21c0ba58330916e7d9 defined\nthe basepython in common place but\ndid not set the ignore_basepython_conflict to True\nin tox to avoid the wrong version in py env and\nerror in future.\n\nChange-Id: I398df150ebf6e5ec9bc34344f0f95a6eb2d0244d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/809e959b144244f449f55b8178ca5bd403a0192c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/809e959b144244f449f55b8178ca5bd403a0192c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
