)]}'
{"id":"openstack%2Frenderspec~717572","triplet_id":"openstack%2Frenderspec~master~I38ea03576c19cb008f5a07be3527535b4652a4c5","project":"openstack/renderspec","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I38ea03576c19cb008f5a07be3527535b4652a4c5","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-05 08:26:38.000000000","updated":"2020-04-07 04:36:12.000000000","submitted":"2020-04-07 04:34:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717572-1586234072441-7870f53a","meta_rev_id":"b233a4d1bf76f241da3bfb76c5f0d526306ea64f","_number":717572,"virtual_id_number":717572,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-06 14:24:50.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 04:34:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-06 14:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 04:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 04:26:46.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-04-07 04:34:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8249fbbf7a64861b90328911fd4228a29c04e86b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-05 08:26:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4e0c11b864effdbe71b47d24b8fd112d081947f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 08:31:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9bbd367561f24eb18f263c0084934fce : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28e7a951a44457c88ae00a4b653011f : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c24dc132e9f04573a494a9099968f9d2 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a588d002f68948378efde8fd24c063f9 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa882cd4e874b79a46011e207a1a9f1 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0beb451641b473b87941d6abbd1cbbb : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5749231dddd81af558fcfa2add13e7f36f901832","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-05 08:44:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8dad96b85b41d1f01289e612ac9b267163d3f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 08:52:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d263214c7d5243b4bd0c00705b3d8ed2 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd62a8e23f44af7b3d4df3740bcb065 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a468e170c6d40ceb47b1c1a032eb046 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/601a9fcdb0eb4d42a17cb39755ca1e55 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3db064ff14bb4e8dbccfac9ea579ce53 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490c9e15433d40388bc5294e33be06f6 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"52e3f0243b1647c1b148d4f7424c5904a5d43363","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-04-06 14:02:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba6bda565c239cc64ce5194272b0c72239dddb17","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 14:08:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"195c9e4102c3e0088708edca4386b6743f80357e","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-04-06 14:23:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"151f57f00aae7e69fc1628daacae76e19d72ca04","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 14:24:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"935f0f0d09cb691ba9b9613d2daa68b6b7d12764","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-04-06 14:38:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9839e72fb2534d849f7ad95a0176779a83f54ea2","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2020-04-07 04:26:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"211d3cff76a96cc269719dfb2063456635b4fa8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:26:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61da057c68a89318bd79906bcb0461718c63fc80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:34:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7453ec36f109404d9fbde17c762fa4c0 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6016a16f7284eb2b383354e814929e2 : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f7198e9be7a4cb89dfc3eb78caae14f : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a92bb1e021ff4ab6ba2f6eacb536dba2 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341c59ddfcc741899c766de5c9b6be9a : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9ccf53b443eb1e1f7cd21bb033bde122f913adc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:34:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"309548e930a3877b89fd00bb381d5df9851b3a50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 04:36:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddf6979be2441a9b07294323282b68a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ece74aaa98a03ea2f93d6c2762e7961ad299ab00","revisions":{"a9388bb21e124937bbd97a9d0426b57ab14ee658":{"kind":"REWORK","_number":1,"created":"2020-04-05 08:26:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/72/717572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/72/717572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/72/717572/1"}}},"commit":{"parents":[{"commit":"19108775bea5695a360e554e4bbf6f855e16d735","subject":"Python 3.8 tests are passing, declare as supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/19108775bea5695a360e554e4bbf6f855e16d735"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 08:25:31.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 08:26:34.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I38ea03576c19cb008f5a07be3527535b4652a4c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/a9388bb21e124937bbd97a9d0426b57ab14ee658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/a9388bb21e124937bbd97a9d0426b57ab14ee658"}]},"branch":"refs/heads/master"},"ece74aaa98a03ea2f93d6c2762e7961ad299ab00":{"kind":"REWORK","_number":2,"created":"2020-04-05 08:44:03.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/72/717572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/renderspec","ref":"refs/changes/72/717572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/renderspec refs/changes/72/717572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/renderspec refs/changes/72/717572/2"}}},"commit":{"parents":[{"commit":"19108775bea5695a360e554e4bbf6f855e16d735","subject":"Python 3.8 tests are passing, declare as supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/19108775bea5695a360e554e4bbf6f855e16d735"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 08:25:31.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 08:43:59.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I38ea03576c19cb008f5a07be3527535b4652a4c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/ece74aaa98a03ea2f93d6c2762e7961ad299ab00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/renderspec/commit/ece74aaa98a03ea2f93d6c2762e7961ad299ab00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
