)]}'
{"id":"openstack%2Freno~567559","triplet_id":"openstack%2Freno~master~Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229","project":"openstack/reno","branch":"master","topic":"story/1565544","hashtags":[],"change_id":"Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229","subject":"Scan all branches for cache generation","status":"MERGED","created":"2018-05-10 10:54:41.000000000","updated":"2019-10-28 14:33:21.000000000","submitted":"2019-10-28 14:29:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"567559-1572272964263-df6c426c","meta_rev_id":"5c8a90a1a7c1c16f8cdcb9a3c1bb966d3b94867c","_number":567559,"virtual_id_number":567559,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 14:29:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-08 21:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-10-28 14:11:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-10-28 14:11:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-10 20:19:07.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2018-09-21 15:10:10.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-06 19:37:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-08 21:38:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 14:11:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-28 14:29:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"696ec4005103930ab6b63ec9332455f588f8c7a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-10 10:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e92964d6c552666920e43bf00fed58578466a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-10 11:06:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/567559/1/check/openstack-tox-pep8/0d8cd9b/ : SUCCESS in 2m 27s\n- openstack-tox-py27 http://logs.openstack.org/59/567559/1/check/openstack-tox-py27/6f523aa/ : SUCCESS in 2m 56s\n- openstack-tox-py35 http://logs.openstack.org/59/567559/1/check/openstack-tox-py35/75f1ec6/ : SUCCESS in 2m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/567559/1/check/build-openstack-sphinx-docs/1d7ad69/ : FAILURE in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f55eae945dbea44e29be98179d78495912002146","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-05-10 14:48:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a1d2b0a3d2837f85339990615f0ca1c105fe2205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-10 14:58:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/567559/2/check/openstack-tox-pep8/8de4829/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/59/567559/2/check/openstack-tox-py27/eb98dd9/ : SUCCESS in 2m 58s\n- openstack-tox-py35 http://logs.openstack.org/59/567559/2/check/openstack-tox-py35/d1b40e5/ : SUCCESS in 3m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/567559/2/check/build-openstack-sphinx-docs/478e727/ : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"bb60dc37bde234001c9f77fcdf6e4a3e1db5592d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-10 20:19:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"859d5d8bd54b65fd7e0139a3e0cc641552e1a71d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-11 21:13:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d49044afa2aef4b98e2faf71ee701beeb7283811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-11 21:20:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/567559/3/check/openstack-tox-pep8/8bc4701/ : SUCCESS in 2m 07s\n- openstack-tox-py27 http://logs.openstack.org/59/567559/3/check/openstack-tox-py27/bca2617/ : SUCCESS in 2m 46s\n- openstack-tox-py35 http://logs.openstack.org/59/567559/3/check/openstack-tox-py35/f8f2ead/ : SUCCESS in 2m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/59/567559/3/check/build-openstack-sphinx-docs/123f281/ : FAILURE in 2m 07s","accounts_in_message":[],"_revision_number":3},{"id":"eb83e091fe458a0a539b0ae1d2a591d5a795164a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-06 19:37:57.000000000","message":"Patch Set 3:\n\nStephen, want to update this and fix the docs?","accounts_in_message":[],"_revision_number":3},{"id":"9740cc31edadae23bf25505d42085b4fe76246f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-07 14:37:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c90f1f3d1df6211563839beded8e574b6f228eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:52:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe73724e5756492bb126a46b588a47c8 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a155ed0555e140b28b5730c777dbd67b : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1276516793f4d48b409c47317670088 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f12352708cb3415b99de6fca289d4581 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/086e9e47be664bf09d088e0ae50062de : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ce322d198d4946b6d6145e8eeee70a : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3662b57eab346c493815093f2135f92 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":4},{"id":"9b8f6f83290ef763510c99cadfaf40442c64a00f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 21:38:18.000000000","message":"Patch Set 4: Code-Review+2\n\nI just rebased the patch, so +2 from me on Stephen\u0027s changes.","accounts_in_message":[],"_revision_number":4},{"id":"885c6b8a3f61868101de4ddd7232277ac9e3a515","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-10 08:54:40.000000000","message":"Patch Set 4:\n\nLGTM but I\u0027d rather not change anything in the toolchain one week before release... Let\u0027s merge that after Train release.","accounts_in_message":[],"_revision_number":4},{"id":"1c968a9563e6b0e81f54968fef7b644adafdd9ec","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-28 14:11:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7595dfade80951a3d9c3d3f98f0c91596f68195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:11:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"509b9839cb8a8fa02edb69b104c27553fe4f0833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:29:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a972fae912cf451daa152222d92da054 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38de62837c634412b85fd5f095f2b0d5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d405979bef45c483956ec47fd7295b : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bff777e0f8041cf8830636e19da62a3 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2635fb40bf6475e8cb18ac86fa4ee97 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3f7500cf0494f3d942cfe891264de0e : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"e9880a9ba0fd4d8b017abb6a3c7c4dd4c932d46b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:29:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2c3163296a7ee07d57a0a46a54eef802c471522c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:33:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f28888a1346471e983c1278cec49d3c : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4b2de693544e039bb2e28a2ffa518c613839c97b","revisions":{"fa13b251d1acf509a13befeae0c79d2f65bba980":{"kind":"REWORK","_number":1,"created":"2018-05-10 10:54:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/567559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/59/567559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/59/567559/1"}}},"commit":{"parents":[{"commit":"79e017ccd051120906d50cec6f813129bbf51de2","subject":"tests: Use mock decorator instead of context manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/79e017ccd051120906d50cec6f813129bbf51de2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 09:37:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 10:54:01.000000000","tz":60},"subject":"Scan all branches for cache generation","message":"Scan all branches for cache generation\n\nTo date, both the \u0027reno cache\u0027 tool and related distutils command only\nscanned the current branch. This resulted in an incomplete cache for\nprojects that used multiple branches. Resolve this by including info\nfrom all enabled branches.\n\nChange-Id: Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: #1565544\nTask: #14465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/fa13b251d1acf509a13befeae0c79d2f65bba980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/fa13b251d1acf509a13befeae0c79d2f65bba980"}]},"branch":"refs/heads/master"},"51c0d9b7e08c7661c06034091d2b35a6143128e6":{"kind":"REWORK","_number":2,"created":"2018-05-10 14:48:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/567559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/59/567559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/59/567559/2"}}},"commit":{"parents":[{"commit":"79bebc4d15bef06c77d5ad550776b1b866d2b43d","subject":"tests: Use mock decorator instead of context manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/79bebc4d15bef06c77d5ad550776b1b866d2b43d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 09:37:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 12:53:50.000000000","tz":60},"subject":"Scan all branches for cache generation","message":"Scan all branches for cache generation\n\nTo date, both the \u0027reno cache\u0027 tool and related distutils command only\nscanned the current branch. This resulted in an incomplete cache for\nprojects that used multiple branches. Resolve this by including info\nfrom all enabled branches.\n\nChange-Id: Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: #1565544\nTask: #14465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/51c0d9b7e08c7661c06034091d2b35a6143128e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/51c0d9b7e08c7661c06034091d2b35a6143128e6"}]},"branch":"refs/heads/master"},"18d0d25ca9db830dff825e46c20a0ce894449e96":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-05-11 21:13:02.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/59/567559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/59/567559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/59/567559/3"}}},"commit":{"parents":[{"commit":"0217f4c08d7ca3a3b8e34af877e90a7e5b0a15ce","subject":"report line numbers for generated content more accurately","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/0217f4c08d7ca3a3b8e34af877e90a7e5b0a15ce"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 09:37:20.000000000","tz":60},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-05-11 21:12:55.000000000","tz":-240},"subject":"Scan all branches for cache generation","message":"Scan all branches for cache generation\n\nTo date, both the \u0027reno cache\u0027 tool and related distutils command only\nscanned the current branch. This resulted in an incomplete cache for\nprojects that used multiple branches. Resolve this by including info\nfrom all enabled branches.\n\nChange-Id: Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: #1565544\nTask: #14465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/18d0d25ca9db830dff825e46c20a0ce894449e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/18d0d25ca9db830dff825e46c20a0ce894449e96"}]},"branch":"refs/heads/master"},"4b2de693544e039bb2e28a2ffa518c613839c97b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-07 14:37:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/567559/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/59/567559/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/59/567559/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/59/567559/4"}}},"commit":{"parents":[{"commit":"8d94d16714c304ac415b876248d5cb7062c3561c","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/8d94d16714c304ac415b876248d5cb7062c3561c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-05-10 09:37:20.000000000","tz":60},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-07 14:37:51.000000000","tz":0},"subject":"Scan all branches for cache generation","message":"Scan all branches for cache generation\n\nTo date, both the \u0027reno cache\u0027 tool and related distutils command only\nscanned the current branch. This resulted in an incomplete cache for\nprojects that used multiple branches. Resolve this by including info\nfrom all enabled branches.\n\nChange-Id: Ibe92a9fa5f42b5a7160201cd64ee6a93f9a78229\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: #1565544\nTask: #14465\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4b2de693544e039bb2e28a2ffa518c613839c97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4b2de693544e039bb2e28a2ffa518c613839c97b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
