)]}'
{"id":"openstack%2Freno~670683","triplet_id":"openstack%2Freno~master~If9ace27e39dfe512feb8e96546bf4e4dbadebd65","project":"openstack/reno","branch":"master","topic":"fix-sphinx","hashtags":[],"change_id":"If9ace27e39dfe512feb8e96546bf4e4dbadebd65","subject":"Fix py27 test: Update sphinx requirement","status":"MERGED","created":"2019-07-13 05:28:07.000000000","updated":"2019-07-20 09:30:29.000000000","submitted":"2019-07-20 09:30:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670683-1563615029197-cc061412","meta_rev_id":"09ba5a3a7c53a9b7d6e496d00c8f84278805138a","_number":670683,"virtual_id_number":670683,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-20 09:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-13 19:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-20 09:24:04.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-14 02:21:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-07-13 19:43:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-20 09:24:04.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-13 19:30:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-13 19:43:38.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-14 02:21:40.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-07-20 09:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b247697690715fc2ad4f1c8d26f3c4c56885ef4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-13 05:28:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5a79ab709675033a1fe14393333958179ff5e6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 05:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/83/670683/1/check/test-release-openstack/61f3433/ : SUCCESS in 1m 42s\n- openstack-tox-pep8 http://logs.openstack.org/83/670683/1/check/openstack-tox-pep8/762dd9d/ : SUCCESS in 2m 21s\n- openstack-tox-py27 http://logs.openstack.org/83/670683/1/check/openstack-tox-py27/ee0e594/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/83/670683/1/check/openstack-tox-py35/d740bbe/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/83/670683/1/check/openstack-tox-py36/cc7996a/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/83/670683/1/check/openstack-tox-py37/2be7888/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/83/670683/1/check/openstack-tox-cover/44c0f42/cover/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/83/670683/1/check/openstack-tox-docs/f2813dc/html/ : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/670683/1/check/openstack-tox-lower-constraints/7c15e6b/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6659450b40d5a2c5741974f01121bbac7687857a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-13 19:30:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b12bf3f8c1d4fa0c1ee3f5cbd8992465b33eb005","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 19:43:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd444e725464149a7be00edd0729aa06c482cc5a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-07-14 02:21:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ec7a0b2bd1b65f1aa001a9ffded3ab32364f9da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-20 09:24:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nsingle core approving to move us forward","accounts_in_message":[],"_revision_number":1},{"id":"5002c015f20043a914b2047dfb7be5e19ad00465","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:24:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ad06d52d44334c689c33a14e54430b46b74d281","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:30:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/83/670683/1/gate/test-release-openstack/045a791/ : SUCCESS in 2m 10s\n- openstack-tox-pep8 http://logs.openstack.org/83/670683/1/gate/openstack-tox-pep8/b789b89/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/83/670683/1/gate/openstack-tox-py27/c4799d5/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/83/670683/1/gate/openstack-tox-py35/ac4db8f/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/83/670683/1/gate/openstack-tox-py36/014b1b7/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/83/670683/1/gate/openstack-tox-py37/aedccab/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/83/670683/1/gate/openstack-tox-docs/a09c969/html/ : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/670683/1/gate/openstack-tox-lower-constraints/11f9ea1/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cd86757ff5ea7334f03a62b5fcc1ae0d15565fde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:30:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7a00fb5a276e9a016d79d5f61d771ead66892ed","revisions":{"e7a00fb5a276e9a016d79d5f61d771ead66892ed":{"kind":"REWORK","_number":1,"created":"2019-07-13 05:28:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/83/670683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/83/670683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/83/670683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/83/670683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/83/670683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/83/670683/1"}}},"commit":{"parents":[{"commit":"51a8f419cac69126eab184874e36eded5312d6ca","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/51a8f419cac69126eab184874e36eded5312d6ca"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-13 05:27:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-13 05:27:04.000000000","tz":120},"subject":"Fix py27 test: Update sphinx requirement","message":"Fix py27 test: Update sphinx requirement\n\npy27 test is failing since sphinx 2.1 does not work with python 2.7, cap\nsphinx requirements for python 2.7.\n\nChange-Id: If9ace27e39dfe512feb8e96546bf4e4dbadebd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e7a00fb5a276e9a016d79d5f61d771ead66892ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e7a00fb5a276e9a016d79d5f61d771ead66892ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
