)]}'
{"id":"openstack%2Freno~672200","triplet_id":"openstack%2Freno~master~Ie03eb805ddd63b40c56d83b523c8d40d0527851f","project":"openstack/reno","branch":"master","topic":"docutils","hashtags":[],"change_id":"Ie03eb805ddd63b40c56d83b523c8d40d0527851f","subject":"Avoid docutils 0.15 with py2","status":"MERGED","created":"2019-07-23 04:49:33.000000000","updated":"2019-07-23 13:45:07.000000000","submitted":"2019-07-23 13:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672200-1563889507550-6f1d36b5","meta_rev_id":"bca237bf00c2e07159b014b953efd7b8547dcf45","_number":672200,"virtual_id_number":672200,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-07-23 13:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-23 13:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-23 13:29:38.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-07-23 13:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 13:29:38.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-07-23 13:33:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-23 13:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61d5b359475c0da7526d4f2bbfc367b0757813a2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-23 04:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fad18e65398f120b22bf35b79d8d05a27013e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 05:01:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/00/672200/1/check/test-release-openstack/c35e82c/ : SUCCESS in 1m 56s\n- openstack-tox-pep8 http://logs.openstack.org/00/672200/1/check/openstack-tox-pep8/f21e2f8/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/00/672200/1/check/openstack-tox-py27/5bfe889/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/00/672200/1/check/openstack-tox-py35/f86c8e6/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/00/672200/1/check/openstack-tox-py36/cf8ce2a/ : SUCCESS in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/00/672200/1/check/openstack-tox-py37/75f103e/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/00/672200/1/check/openstack-tox-cover/f1d7252/cover/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/00/672200/1/check/openstack-tox-docs/fe1707a/html/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/672200/1/check/openstack-tox-lower-constraints/3cb8492/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c2342e264cc0c11cc87ae86391a7e5751fb659f1","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-07-23 13:29:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3496b40b8b3a8ef8dff8ff2af278b0f7af103aec","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-23 13:33:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d40ee74d4cedb50a679eaa3533937cf9dfee8e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:34:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"138f56e18ce2a146dce3dcc5b446c8fca380f64e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:45:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08154f4fa47045dab335cd4ddff39368c2290600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:45:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/00/672200/1/gate/test-release-openstack/822b5d9/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/00/672200/1/gate/openstack-tox-pep8/6f6987a/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/00/672200/1/gate/openstack-tox-py27/2551618/ : SUCCESS in 6m 57s\n- openstack-tox-py35 http://logs.openstack.org/00/672200/1/gate/openstack-tox-py35/8a33d85/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/00/672200/1/gate/openstack-tox-py36/ef75fe6/ : SUCCESS in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/00/672200/1/gate/openstack-tox-py37/4a9bbeb/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/00/672200/1/gate/openstack-tox-docs/ee403c1/html/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/672200/1/gate/openstack-tox-lower-constraints/b3e5ac1/ : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd9d7924025800213c472cc053df32f57ee39825","revisions":{"dd9d7924025800213c472cc053df32f57ee39825":{"kind":"REWORK","_number":1,"created":"2019-07-23 04:49:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/672200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/00/672200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/00/672200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/00/672200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/00/672200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/00/672200/1"}}},"commit":{"parents":[{"commit":"e7a00fb5a276e9a016d79d5f61d771ead66892ed","subject":"Fix py27 test: Update sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e7a00fb5a276e9a016d79d5f61d771ead66892ed"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-23 04:47:09.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-23 04:49:07.000000000","tz":120},"subject":"Avoid docutils 0.15 with py2","message":"Avoid docutils 0.15 with py2\n\ndocutils 0.15 breaks the py2 tests, avoid the version for py2.\n\nChange-Id: Ie03eb805ddd63b40c56d83b523c8d40d0527851f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/dd9d7924025800213c472cc053df32f57ee39825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/dd9d7924025800213c472cc053df32f57ee39825"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
