)]}'
{"id":"openstack%2Freno~705674","triplet_id":"openstack%2Freno~master~I1a3b3e542f2a6cb03f98d80ec5685c315031eed7","project":"openstack/reno","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I1a3b3e542f2a6cb03f98d80ec5685c315031eed7","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2020-02-04 13:29:50.000000000","updated":"2020-02-05 13:21:49.000000000","submitted":"2020-02-05 13:16:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705674-1580908607019-bbef5f2b","meta_rev_id":"e59badc75a01c57b3f045e58a616925b999675db","_number":705674,"virtual_id_number":705674,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 13:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-02-05 13:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-02-04 20:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2020-02-04 18:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-02-05 13:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 18:23:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-04 20:28:01.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-02-05 13:06:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-05 13:16:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12d83a007db44361797ae3626b057e053714e6d6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 13:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ada05b10a1d9cce1079a93377bbf1108358a4286","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 13:32:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8282c0d89ce50a2e1ee00b8a6c50917bb36d0a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:45:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd1c21a55de346959dfe89b51239dfb0 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344712c6f0eb4e5a932b36fccd10be62 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77eb0ac0001d4a51bedb5dcbec9b3c10 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edd87169cde2497e84bbf18d95c48542 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c5a00cdbde4405eafbd7d2d0433942e : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbe4cc1788394817a001fa5b36284eb8 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a434acd0e6433791f86afc5b6cfb17 : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87774b29b9b84094b02bb7de8004053d : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"870a35c96d78a1f7a87aac24c5b9aca2d1f2dbbd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 18:23:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"076dca10d183a96330a7cf71dda2d987efc2cf21","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-02-04 20:28:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05b243d322ef8f9fcc6fe8b3806cafd4a7035f45","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-05 13:06:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98972edacb04907070e6c2e4bb1555b78b94e61e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:06:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7288647af35b1052aec9cf5b5c41a31fedefd56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:16:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/797ea9ca77fb4c1f96120f0ccfe21d2e : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87edbc4d1424fe8ad0cef68fb9ba131 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2101de0edd45418e9ed192f7e9d4b776 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4402d25e10224e83927e2cc473d7bbda : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3ee3abb8604741b3facb57935efa8f : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77f411613735466a9957f82435b012ee : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2a5a87cfe55c839a50b1e202d90c8137f1ef1471","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:16:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8c6f15f3aac66e1e3414a3ab8235bfee239c0a84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:21:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def6c0f9f26641d4ad361802376ce47d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"60699c1b06ae39ba2e98548fdc2c16b3da37d642","revisions":{"26b3b2bf2c9a115ec7006aaccae693fda29c679a":{"kind":"REWORK","_number":1,"created":"2020-02-04 13:29:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/705674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/74/705674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/74/705674/1"}}},"commit":{"parents":[{"commit":"b02339be5fbe5e5921162e574879c50ed4011075","subject":"Merge \"Scan all branches for cache generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/b02339be5fbe5e5921162e574879c50ed4011075"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 13:23:29.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 13:28:11.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nRemove install_command from tox.ini as cleanup.\n\nChange-Id: I1a3b3e542f2a6cb03f98d80ec5685c315031eed7\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/26b3b2bf2c9a115ec7006aaccae693fda29c679a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/26b3b2bf2c9a115ec7006aaccae693fda29c679a"}]},"branch":"refs/heads/master"},"60699c1b06ae39ba2e98548fdc2c16b3da37d642":{"kind":"REWORK","_number":2,"created":"2020-02-04 13:32:42.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/705674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/74/705674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/74/705674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/74/705674/2"}}},"commit":{"parents":[{"commit":"b02339be5fbe5e5921162e574879c50ed4011075","subject":"Merge \"Scan all branches for cache generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/b02339be5fbe5e5921162e574879c50ed4011075"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 13:23:29.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 13:32:27.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nRemove install_command from tox.ini as cleanup.\nSwitch to Ussuri jobs\n\nChange-Id: I1a3b3e542f2a6cb03f98d80ec5685c315031eed7\nSem-Ver: api-break\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/60699c1b06ae39ba2e98548fdc2c16b3da37d642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/60699c1b06ae39ba2e98548fdc2c16b3da37d642"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
