)]}'
{"id":"openstack%2Freno~716666","triplet_id":"openstack%2Freno~master~Ic6fe4ec404b02aee398b291a9b1430b6700aa9c6","project":"openstack/reno","branch":"master","topic":"configurable-scanning","hashtags":[],"change_id":"Ic6fe4ec404b02aee398b291a9b1430b6700aa9c6","subject":"Allow configuration via environment variables","status":"ABANDONED","created":"2020-04-01 16:49:40.000000000","updated":"2020-04-02 15:42:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"694270c57a9e6910d1f208f09a56dd6adfa83380","_number":716666,"virtual_id_number":716666,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"date":"2020-04-01 19:00:24.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-01 17:09:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2020-04-01 18:13:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 17:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 19:00:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"7b899f7e374b632d68fc0a91eb9f6e9301b952c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-01 16:49:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41d2fc50b16ddfbd80f7cf191ccac396d82802fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 17:09:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6694f066734c719e185daf7f024fac : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00fc66600f5d498f98f7f187f79deecf : FAILURE in 9m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d50ee7604d5e479aad03413c516cb144 : FAILURE in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef88a8bb342e4e39b2783458874a91c1 : FAILURE in 7m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf5930d177594a5f82c061313d24bb53 : FAILURE in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5d0b00b0194ba293316170f281f10c : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94983f40ecd446298be8284c981e6196 : FAILURE in 7m 51s","accounts_in_message":[],"_revision_number":1},{"id":"82a44b74ebaec1c4ac12892f428de3e7ef6f0ed1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-01 18:13:38.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds more work","accounts_in_message":[],"_revision_number":1},{"id":"502d2c4080518d8bba251cb207abe8cae69d3a9e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-01 18:58:53.000000000","message":"Patch Set 1:\n\nThis test [1] should append \"Environment: ``RENO_RELEASE_TAG_RE``\" to the expected string used on assert.\n\n[1] https://github.com/openstack/reno/blob/master/reno/tests/test_exts.py#L124","accounts_in_message":[],"_revision_number":1},{"id":"ac2f5a78bef415b0727cd379a7862be4e0a87efa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-01 19:00:24.000000000","message":"Patch Set 1:\n\n\u003e This test [1] should append \"Environment: ``RENO_RELEASE_TAG_RE``\"\n \u003e to the expected string used on assert.\n \u003e \n \u003e [1] https://github.com/openstack/reno/blob/master/reno/tests/test_exts.py#L124\n\nSorry, I meant we need to append \"Environment: ``RENO_RELEASE_TAG_RE``\" to the string.","accounts_in_message":[],"_revision_number":1},{"id":"9ce00fa531d525b59c7a3233de914bfb835256c6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-02 15:42:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a927c78660b19f04d0a3c20662a767376b4cb93","revisions":{"2a927c78660b19f04d0a3c20662a767376b4cb93":{"kind":"REWORK","_number":1,"created":"2020-04-01 16:49:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/716666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/66/716666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/66/716666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/66/716666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/66/716666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/66/716666/1"}}},"commit":{"parents":[{"commit":"ab3eb972bfb66b36f7979e071e8157150d5ecc85","subject":"Remove usage of six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ab3eb972bfb66b36f7979e071e8157150d5ecc85"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-01 16:48:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-01 16:48:40.000000000","tz":60},"subject":"Allow configuration via environment variables","message":"Allow configuration via environment variables\n\nThe pbr integration of reno doesn\u0027t provide an easy way to configure\nreno itself. Make this possible using environment variables.\n\nChange-Id: Ic6fe4ec404b02aee398b291a9b1430b6700aa9c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/2a927c78660b19f04d0a3c20662a767376b4cb93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/2a927c78660b19f04d0a3c20662a767376b4cb93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
