)]}'
{"id":"openstack%2Freno~718229","triplet_id":"openstack%2Freno~master~I9c76003609b8d5ca4ca2a78db96e66597eb5a6bb","project":"openstack/reno","branch":"master","topic":"nonlocal-uri","hashtags":[],"change_id":"I9c76003609b8d5ca4ca2a78db96e66597eb5a6bb","subject":"do not include external svgs from readme in docs","status":"MERGED","created":"2020-04-07 21:14:53.000000000","updated":"2020-04-07 21:43:10.000000000","submitted":"2020-04-07 21:40:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718229-1586295622301-49baf0dc","meta_rev_id":"65ef5c0e260e0f86ef7720046c21fca498ba14c5","_number":718229,"virtual_id_number":718229,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 21:40:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 21:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-07 21:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 21:22:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 21:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46c97c526ad587241917a26361e1c97fb28f1550","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-04-07 21:14:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0f7d12a85e24e847dadbf4eefc45bf8d8237b11","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 21:22:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTested locally. Docs index page looks better like this, in addition to avoiding the external inclusion of project metadata.","accounts_in_message":[],"_revision_number":1},{"id":"872bb2c51843630d93d42b17ff83518feb941b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea8c753e51964096900a28a0c749dfc6 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/145fb7bf19bd470187adaffff8ad64eb : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e58aa408748888026f17a7ce41aaa9f6b94b6496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:27:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a9f6c1aa6edbe46c1568adb8a07d98a8b435f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:40:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a0775d296b4534a6e8b7b65ea34775 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e8e3ea645149f7a3f98f6f506909ac : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":1},{"id":"91b20e9bda1ccb102b8cba3d6027165e366d9438","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:40:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e55329923deb360fa2b0a8e342a49ea692f38ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:43:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c121781e4eb04ae6935c5e76fbe60f4f : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ad655bc726b15214e1d66fc832d82a2012e6e9c","revisions":{"6ad655bc726b15214e1d66fc832d82a2012e6e9c":{"kind":"REWORK","_number":1,"created":"2020-04-07 21:14:53.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/718229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/29/718229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/29/718229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/29/718229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/29/718229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/29/718229/1"}}},"commit":{"parents":[{"commit":"bd3886ac480500eb7515058cfe2ea5662fa53d8d","subject":"scan the whole branch if a version cannot be found","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/bd3886ac480500eb7515058cfe2ea5662fa53d8d"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-04-07 21:05:15.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-04-07 21:05:15.000000000","tz":-240},"subject":"do not include external svgs from readme in docs","message":"do not include external svgs from readme in docs\n\nOffline documentation builds shouldn\u0027t include the SVGs, so don\u0027t\ninclude that section of the readme in the HTML docs.\n\nChange-Id: I9c76003609b8d5ca4ca2a78db96e66597eb5a6bb\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/6ad655bc726b15214e1d66fc832d82a2012e6e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/6ad655bc726b15214e1d66fc832d82a2012e6e9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
