)]}'
{"id":"openstack%2Freno~719950","triplet_id":"openstack%2Freno~master~Ic0ed1305cf21deb6abea0991ec604d7f9b821622","project":"openstack/reno","branch":"master","topic":"trivial","hashtags":[],"change_id":"Ic0ed1305cf21deb6abea0991ec604d7f9b821622","subject":"Keep scanning master branch after first shared commit","status":"MERGED","created":"2020-04-14 13:07:36.000000000","updated":"2020-04-27 19:27:32.000000000","submitted":"2020-04-27 19:24:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719950-1588015458586-54f1cd87","meta_rev_id":"7340dacbf7fb2584f8dce117c6cf230de9ce5f44","_number":719950,"virtual_id_number":719950,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 19:24:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 19:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-23 10:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 19:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 19:19:30.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-04-23 10:23:32.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-27 19:11:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-27 19:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc52095a87e195b31067a50049bff5ca4e603cee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 13:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0080af900d1955f9219d9670e24a38c5553e196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:41:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d105f30d40d646e0a076964fbaa8421a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56ab9bdd833f4496ac395d0a78d2f240 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b15002bcdfa4b7eb4382172520f5a4f : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae7a5a29d97a44fb832fbd7501f094f7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db569bae8e58465e9ad9818713d5923d : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94405dfde4434bab800e506d0cd1e017 : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e39d43f7dd0a45818e99c24e6b3b081e : FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2f71a79bdd5ae814a535de377e0d080b12bc326a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-04-14 13:46:58.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI think this is probably safe. One comment on the testing, inline. We will need a good release note describing the change in behavior.\n\nOn IRC I also proposed some manual testing against nova and other projects with longer history.","accounts_in_message":[],"_revision_number":1},{"id":"e5551ff69a1926a50d435a20ae9584c82b01852b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 16:24:54.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e I think this is probably safe. One comment on the testing, inline.\n \u003e We will need a good release note describing the change in behavior.\n\nComing up shortly.\n\n \u003e On IRC I also proposed some manual testing against nova and other\n \u003e projects with longer history.\n\nDid this testing against nova. No changes.\n\n  $(.venv) diff ../nova-reno.cache.before ../nova-reno.cache.after\n  $(.venv)\n\n(\u0027before\u0027 was generated using head of master and \u0027after\u0027 using PS1)\n\n  $(.venv) pip freeze\n  certifi\u003d\u003d2020.4.5.1\n  dulwich\u003d\u003d0.19.15\n  pbr\u003d\u003d5.4.5\n  PyYAML\u003d\u003d5.3.1\n  -e git+git@github.com:openstack/reno.git@441c9669d2bda650030873a6b129153e6cd3fbd8#egg\u003dreno\n  urllib3\u003d\u003d1.25.8","accounts_in_message":[],"_revision_number":1},{"id":"1d52ce3f5dcd43e97f0aefbfd0ca346153bae2a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 16:36:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2e685efa61500b24ab5c14be98ccbfbc4479be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 16:51:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5070b97dc50348508b55df4e899da90d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e05c021e4e264c918e0a40734aeff335 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a0302a7ee8b48f7b69a6a36fb26cce2 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91a969776eb4cf7bcb38fd414731f37 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac44a41d601b4de0836f20584a9d0652 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0a9f0e0349486dbc60dc9b3408552a : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/924addbeaa4347a2b98471cce5001ae9 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e219149aa9d14d6fe55a71693f63ff98959f99a1","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-04-14 19:19:30.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nA couple of typos in the release note, but otherwise this looks good.","accounts_in_message":[],"_revision_number":2},{"id":"c533957172efbf49c8cbee3781c3cb48e52b6868","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-21 10:35:29.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"144137985ccae7e36249aef5e358a662fe53eddc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-21 10:35:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e0a4936a4aa0d636c9c601e8a413c03e9bf029cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 10:46:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1461131af1c044fd910675055283c844 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8709f6a9ef42493489c6a0092a89155b : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca52eacbcb8947bca216b21b89fb22b7 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e617e728184042308a878d2ce25aed79 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8ea6aa70d4400e8b9104d168cac53c : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb8ce3257de477d9e92141b4475db19 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9205048cfe4549ba94684f6ebbcf4d59 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3},{"id":"90d354e50fc14289795e0a99be0347920881870b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-23 10:23:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1b864450d04c9f9035b0b635381bbc834ac1062b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 18:17:49.000000000","message":"Patch Set 3:\n\nI don\u0027t see how it could be related, but if I remove the reno setting closed_branch_tag_re in the Cinder repo, I get the following:\n\n  File \"/home/smcginnis/src/openstack/reno/reno/sphinxext.py\", line 67, in run\n    conf \u003d config.Config(reporoot, relnotessubdir)\n  File \"/home/smcginnis/src/openstack/reno/reno/config.py\", line 209, in __init__\n    self._load_file()\n  File \"/home/smcginnis/src/openstack/reno/reno/config.py\", line 231, in _load_file\n    self.override(**self._contents)\nTypeError: override() argument after ** must be a mapping, not NoneType\n\nGoing to check some things first, but I don\u0027t think this is a result of these changes.","accounts_in_message":[],"_revision_number":3},{"id":"a13483f040ec1c8e33f9f7ab39bc7100a9dce703","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 19:11:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nYep, not related.","accounts_in_message":[],"_revision_number":3},{"id":"ce173a1070aa14ba791f1fb299c5905568126d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:11:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7157e9b1bb445d7afbcb89a315d4f4dd888ea76d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:24:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dbc59abdae8468493251dafc3ba3399 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce60b7b268ff430982e1302e37decab8 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ad5effba3654f0e89a9b1ab32181ece : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b23d5fdc454b07b7440cb9d0c5983c : SUCCESS in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eec67636f8654a9f8a7bb2c96c314622 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":3},{"id":"0c338ec56858fe9cb65b553d03918e8156b3f2c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:24:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"86184a93eb54f7cd393aa4c8a4ee5beeb8dbb14d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:27:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff715aa93ea473dbaaca37f3fc0e34f : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"370f1a5c7b0422a331a5ffa48db244c5ddb75cf4","revisions":{"441c9669d2bda650030873a6b129153e6cd3fbd8":{"kind":"REWORK","_number":1,"created":"2020-04-14 13:07:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/719950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/50/719950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/50/719950/1"}}},"commit":{"parents":[{"commit":"deb82aaaa7c6466f3bd89f4a2ebaa67ccfd0cf8a","subject":"tox: Suppress output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/deb82aaaa7c6466f3bd89f4a2ebaa67ccfd0cf8a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-05 20:08:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-14 13:06:44.000000000","tz":60},"subject":"Keep scanning master branch after first shared commit","message":"Keep scanning master branch after first shared commit\n\nIf no earliest version is provided for a branch, reno will scan commits\non the branch in reverse order, attempting to find a common ancestor\nwith the master branch. Currently, the first patch this finds must be\nthe tagged one or it will quit early. There\u0027s no particular reason to do\nthis. Instead, we can keep going for a little longer. This is helpful\nfor situations where we haven\u0027t branched off straight after the tag.\n\nChange-Id: Ic0ed1305cf21deb6abea0991ec604d7f9b821622\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/441c9669d2bda650030873a6b129153e6cd3fbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/441c9669d2bda650030873a6b129153e6cd3fbd8"}]},"branch":"refs/heads/master"},"9f46306bba2f1a13078ebedf1dbab6854b16e624":{"kind":"REWORK","_number":2,"created":"2020-04-14 16:36:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/719950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/50/719950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/50/719950/2"}}},"commit":{"parents":[{"commit":"4a93566f49a48cfc2c0bb878c26680d73a14820b","subject":"tox: Suppress output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4a93566f49a48cfc2c0bb878c26680d73a14820b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-05 20:08:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-14 16:36:22.000000000","tz":60},"subject":"Keep scanning master branch after first shared commit","message":"Keep scanning master branch after first shared commit\n\nIf no earliest version is provided for a branch, reno will scan commits\non the branch in reverse order, attempting to find a common ancestor\nwith the master branch. Currently, the first patch this finds must be\nthe tagged one or it will quit early. There\u0027s no particular reason to do\nthis. Instead, we can keep going for a little longer. This is helpful\nfor situations where we haven\u0027t branched off straight after the tag.\n\nChange-Id: Ic0ed1305cf21deb6abea0991ec604d7f9b821622\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9f46306bba2f1a13078ebedf1dbab6854b16e624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9f46306bba2f1a13078ebedf1dbab6854b16e624"}]},"branch":"refs/heads/master"},"370f1a5c7b0422a331a5ffa48db244c5ddb75cf4":{"kind":"REWORK","_number":3,"created":"2020-04-21 10:35:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/719950/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/50/719950/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/50/719950/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/50/719950/3"}}},"commit":{"parents":[{"commit":"4a93566f49a48cfc2c0bb878c26680d73a14820b","subject":"tox: Suppress output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4a93566f49a48cfc2c0bb878c26680d73a14820b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-05 20:08:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-04-21 10:35:20.000000000","tz":60},"subject":"Keep scanning master branch after first shared commit","message":"Keep scanning master branch after first shared commit\n\nIf no earliest version is provided for a branch, reno will scan commits\non the branch in reverse order, attempting to find a common ancestor\nwith the master branch. Currently, the first patch this finds must be\nthe tagged one or it will quit early. There\u0027s no particular reason to do\nthis. Instead, we can keep going for a little longer. This is helpful\nfor situations where we haven\u0027t branched off straight after the tag.\n\nChange-Id: Ic0ed1305cf21deb6abea0991ec604d7f9b821622\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/370f1a5c7b0422a331a5ffa48db244c5ddb75cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/370f1a5c7b0422a331a5ffa48db244c5ddb75cf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
