)]}'
{"id":"openstack%2Freno~721336","triplet_id":"openstack%2Freno~master~Ib568709bd01d12ff2fb6ef6fe31b290ea63e7caa","project":"openstack/reno","branch":"master","hashtags":[],"change_id":"Ib568709bd01d12ff2fb6ef6fe31b290ea63e7caa","subject":"Remove future imports","status":"MERGED","created":"2020-04-20 20:07:15.000000000","updated":"2020-04-23 09:29:19.000000000","submitted":"2020-04-23 09:26:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721336-1587634018280-a0c18567","meta_rev_id":"4f5f8f157d97ee512111d31c43dc0b1a797caff9","_number":721336,"virtual_id_number":721336,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 09:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-04-22 06:59:25.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-23 09:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-04-23 09:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-23 09:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 06:59:25.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-23 09:12:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-23 09:13:54.000000000","updated_by":{"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},"reviewer":{"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},"state":"REVIEWER"},{"updated":"2020-04-23 09:26:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"accbc4c4a7e4e993cbde277196ea7e6d7a7cf764","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-20 20:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e626b0476aeb4e14aa2102d94df3a09d18b622b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:20:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a3119b85f44d92b94aa78c3b70a402 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5591dbb0e4f94bcb92c9d292f3454a85 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d329b74cc4d748fea5a4631747e0a760 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09cbb95ddb8f45e1a2d8898e280eb370 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb05c300dd0f4e1295c33d16bb46060e : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc2db59e4ea4e53a560fdfe550552c9 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7f4526856d3418fa6557a70a6a8ed45 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e539028721fb329854ba1a6be2be722e4dafe9c5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 06:29:32.000000000","message":"Patch Set 1:\n\nLet\u0027s merge https://review.opendev.org/#/c/721346/ first - as new contributor, let\u0027s not create a merge conflict directly for them.\n\nI\u0027ll +2 once the other change is in.\n\nthanks!","accounts_in_message":[],"_revision_number":1},{"id":"4939450b9d5d5f97c5b9fa30852f43b92b5c0c9a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:44:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"796a70734a993239c633a5face92c4bc32b640a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:45:02.000000000","message":"Patch Set 2: Code-Review+2\n\nrebased to resolve merge conflict","accounts_in_message":[],"_revision_number":2},{"id":"d0248eaf11a92a9888ecee49d423b62e10b18bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 06:58:35.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89de9aa787874675b9655c4613377031 : FAILURE in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781ee79f23764c23a65dc5538b91a2b8 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aafd191643d4bf292fa8878dcead750 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bff42e7440405aba10b64aa8a84c8d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf75cd587b3145d2b21c317688639c3f : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e56b43a42894a48a77660a446a034a6 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c49bb145851f4590b979022ad5a6eed9 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9196dbc7e09b421c361a834b98bbb234a9fb7dde","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:59:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"772ec1936c194b1637bf5e1aa344413ccffea38c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:59:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7a020e9539e685b8dc7dcab6292f62f470841a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 07:14:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e676bde918b401f8c1fa5ea2502cc41 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/099cf5e1ce2e4ccc8afd2d539227d9c8 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45124cb752b74c84abf7d54381d34e1b : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71b1596e58594cc493d540c091c487c5 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41810a98d9a043218685e0bfbdf378a5 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb317d84d28409898391fe1e7574fae : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3c56b91434545a0a7e774a147d613bd : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":3},{"id":"5b29cf56e816873135a271db3e9ce2b2b486c7a8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-23 09:12:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"90c1b7ec7e43de5e76ce94f60529bee7ff186c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:13:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e09153624497ee1cf5e12a1c4ebd922327f7f21","author":{"_account_id":31849,"name":"dongdong","email":"dudd@awcloud.com","username":"dongdong"},"date":"2020-04-23 09:13:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"40a9da5db1c4506bfb6e348fac745fdc33e868a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:26:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1ef34d9c2bababbafee379b162cc74db337423f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:26:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051ba0a9131b49f68e92ac24dfc137c2 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/020fb9d7b012489582d428a49d4e814a : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2922af84f124a8b82113e02bc87f246 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1425cb037b93410880a6bbd865935e38 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d27ce889bb87487f95238384491993f2 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4f5f8f157d97ee512111d31c43dc0b1a797caff9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:29:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d8cac942e64b3caa5135593c903ccb : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4","revisions":{"0942c8c3ef7b6b040a3256e9817b2539e5683da1":{"kind":"REWORK","_number":1,"created":"2020-04-20 20:07:15.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/721336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/36/721336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/36/721336/1"}}},"commit":{"parents":[{"commit":"a19cd28f20b642cb027c97fed7d93b6e8842dbd8","subject":"sphinxext: Support rootdir discovery without git","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/a19cd28f20b642cb027c97fed7d93b6e8842dbd8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 20:05:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 20:05:41.000000000","tz":-300},"subject":"Remove future imports","message":"Remove future imports\n\nWe only support Python 3 runtimes now. We no longer need future imports.\n\nChange-Id: Ib568709bd01d12ff2fb6ef6fe31b290ea63e7caa\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/0942c8c3ef7b6b040a3256e9817b2539e5683da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/0942c8c3ef7b6b040a3256e9817b2539e5683da1"}]},"branch":"refs/heads/master"},"d8c0a73ff58b1012fbdb6061d68de16b893ab3f2":{"kind":"REWORK","_number":2,"created":"2020-04-22 06:44:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/36/721336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/36/721336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/36/721336/2"}}},"commit":{"parents":[{"commit":"68c19afcb5462d99d1b47e12a73d4b608490e0d8","subject":"Merge \"Add support for release dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/68c19afcb5462d99d1b47e12a73d4b608490e0d8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 20:05:41.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 06:44:12.000000000","tz":120},"subject":"Remove future imports","message":"Remove future imports\n\nWe only support Python 3 runtimes now. We no longer need future imports.\n\nChange-Id: Ib568709bd01d12ff2fb6ef6fe31b290ea63e7caa\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/d8c0a73ff58b1012fbdb6061d68de16b893ab3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/d8c0a73ff58b1012fbdb6061d68de16b893ab3f2"}]},"branch":"refs/heads/master"},"eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4":{"kind":"REWORK","_number":3,"created":"2020-04-22 06:59:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/36/721336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/36/721336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/36/721336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/36/721336/3"}}},"commit":{"parents":[{"commit":"68c19afcb5462d99d1b47e12a73d4b608490e0d8","subject":"Merge \"Add support for release dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/68c19afcb5462d99d1b47e12a73d4b608490e0d8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 20:05:41.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-04-22 06:59:19.000000000","tz":0},"subject":"Remove future imports","message":"Remove future imports\n\nWe only support Python 3 runtimes now. We no longer need future imports.\n\nChange-Id: Ib568709bd01d12ff2fb6ef6fe31b290ea63e7caa\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
