)]}'
{"id":"openstack%2Freno~721728","triplet_id":"openstack%2Freno~master~I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97","project":"openstack/reno","branch":"master","topic":"release-date-cache","hashtags":[],"change_id":"I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97","subject":"Add cache support for release dates","status":"MERGED","created":"2020-04-21 20:56:40.000000000","updated":"2020-04-27 16:31:01.000000000","submitted":"2020-04-27 16:09:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721728-1588003749568-276f548f","meta_rev_id":"0a538c02dece216a65c9c1d4993aba6da71b59e5","_number":721728,"virtual_id_number":721728,"owner":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-25 06:14:39.000000000","_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 16:09:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 15:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},{"value":2,"date":"2020-04-24 14:58:40.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 15:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 14:58:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-27 15:46:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-27 16:09:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"031de7cb963a1860957f1697c08a4c2787f47734","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-21 20:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dd65de8c1961fa941b1402d70eabb0d065cac2b","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-21 20:58:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf8350be640a21dfa99fa440832a266c39960f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 21:06:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41697a7d7b604f7381304bb3fb880578 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f8cacdb58684b12986a8f69cb2052d1 : FAILURE in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bc6b9bba41b4abd9f5f08c7ee20c6f6 : FAILURE in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e4c137c0e114d29831b4cc3a4649d94 : FAILURE in 7m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/727a8b86217849d2be12954e21125569 : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0ba6aa50a54977808b2778262c7344 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9354df9c5f6b4abf9a8ed5644cf00c5a : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9e2fca9329c4cc1190a2d4822af020b90cbc6a56","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-21 22:46:06.000000000","message":"Patch Set 1:\n\nIs there any documentation on how to run the tests locally? Trying to download the logs from Zuul didn\u0027t help as I get a \"The resource could not be found.\" error.","accounts_in_message":[],"_revision_number":1},{"id":"128de0abbbbef1a8ab1ed8a1ac07c18900129328","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-21 23:43:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d7824e020ccd992ff1543e275949ed38ae82929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 23:58:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e788268b7f484e44bb994fd36d41b129 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3435e0b1c20b46a3b4727761e44df3db : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5bcbcd1fa0d407b8f50317c45b34ee8 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fa269066094d71bad69a86826c2d7d : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15c780dbaa274688a2ddf05086d63bf2 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee934488d58445caa8f06e7f2f6b566d : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/808014521bb74c889f9b1a7e0a85c01f : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4b5278ba8ea57cd45e1c17896a5e9abef3d2e331","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 05:47:55.000000000","message":"Patch Set 2:\n\nto run locally the py36 test:\n\ntox -e py36","accounts_in_message":[],"_revision_number":2},{"id":"15d23b74906faf95f6466cdc27f1e930adca7a27","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-22 05:49:18.000000000","message":"Patch Set 2:\n\nThanks Andreas. After some experimentation, I figured it out too.\n\nThis patch is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"269e8ffea0d9b9a99e0b9b03e1439b3ec8dafa60","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 05:50:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7dc9fe6aa03829eb2557528a7f74066fbce7c74d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 05:55:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91ea348ea414d5c0445383695b13a6d35c38a0a5","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-22 05:55:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"555f339ba1e0c3d7ca57ba9436d51a8336e49813","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-22 05:56:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d465949f2a4ff3bad9e7d4265f9e1a87c85d5927","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-22 05:59:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"67220901e48bdbecbf392a6142e45ad9dc260ab6","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-22 06:07:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8c45e386ecbefdc89b0d932ebd563409288394bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 06:07:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1ecd1d93ac4c2ba50470e20bf62f86 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0805475e5a8847f6abbfdaceea8c5990 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5086162c3d7d491da7c30c64d3b741b3 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a600a8fc4f8a4c18ac4750a24460c512 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d4320e61de45baae154c0b3dbb8e7a : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f867fdb06b8543929893040c920eb81e : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08e8586ab8264521810c46a6549ce4f3 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":3},{"id":"8da17a0d313112e36b28bfb4efeddb7a7e138bb1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-23 10:53:04.000000000","message":"Patch Set 3:\n\nthis has now a merge conflict, could you rebase and resolve, please?","accounts_in_message":[],"_revision_number":3},{"id":"0597d19faa421913fe547ea8192efbafc2ad47de","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-23 16:32:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cc18a41ed88e555d336a92c68207b052bd71361d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:46:00.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd71964fecf549fcbae3015fbd7ed9df : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ff4ece3dce4872a833ec890d75bba7 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7be27908392d4e6895254542a1460538 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc39b9fad5548dcb67cc683944fc90f : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225bdf4735e141abb0161bd4cf9cae58 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdfaeebd37f54ab5a5d5ee76de187e55 : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21b80fabfc7e47469e4e30e201d14535 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7c7457912b915891ea6fd07b3d4ff3e5dad6a4c8","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-23 17:13:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"745da8676d9c84cee6620cdd96bb4d8a902f04bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 17:35:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2e31da5b2f4e1289a55ae29ab7c938 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/049347ffc8cd4cb29a92beb612b63396 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e2ee220be924e2a9f97a9b7e25b4d3d : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27084929505f48789e44ce1a062ccba1 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7a8732701b54d20b864312421973195 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d43dc488fd84bec9de4050deaaa2cf6 : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab50e94ce1fb4905b3cbd3752e2d6348 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d64beec8fef7a3a4b68a19dc4e3b0444aab7c3bf","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-23 18:26:53.000000000","message":"Patch Set 5:\n\nAndreas, done.","accounts_in_message":[],"_revision_number":5},{"id":"67b42ae07676c503267fc252a45e3a5fea0a3a96","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-24 14:58:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0b4110c14d33f4cb3ea567ffa2a7cbaabbc49dd9","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-25 06:14:39.000000000","message":"Patch Set 5:\n\nDo I need to do anything else to get this to merge?","accounts_in_message":[],"_revision_number":5},{"id":"eeae652c2f9707c75e90610bdca31cbe5fb0358a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-25 08:06:14.000000000","message":"Patch Set 5:\n\nPlease be patient, this can take a few days for a reviewer to look at this one - it\u0027s release phase for OpenStack. We require a second reviewer.","accounts_in_message":[],"_revision_number":5},{"id":"7cd36d7371851a9f07d94506f884ccd45b5fad70","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-26 00:14:12.000000000","message":"Patch Set 2:\n\nAndreas: My apologies. I wasn\u0027t trying to hurry this patch in but just wasn\u0027t aware of what the OpenStack workflow was to get a patch merge. That you for letting me know that it needs two +2s. Further, I didn\u0027t know it was release time for OpenStack either.\n\nLook forward to someone else giving this a look post-release and once the dust settles. This is not urgent.\n\nWill stay tuned.","accounts_in_message":[],"_revision_number":2},{"id":"b630804e8facb9168479a8d5495ae58d7d2e30a2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 15:46:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nTested locally with one of our repos that have been around for a long time and no issues found.","accounts_in_message":[],"_revision_number":5},{"id":"b5342a3cfac99a0f089c3fdf14662aa13825a58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:46:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2775f6745ea701570559138e7930cac7f68c5941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:09:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe74210a098846ea86ee77844b1cb639 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3974a773ca9a46f797127c07628d536b : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbb591dbc2e94ffd9d8ef40025384181 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55f3ba6d7784614b5fb59567195481d : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26051d546a3a46a8998cb7a7df77a9fb : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":5},{"id":"755d9b88ab4ec1c20aa670b820d944b5beae106a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:09:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3fad9295c45750c9a16ae8036f55797a6b8c6dde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:12:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b813bb075d4548ac3fa0e2a3087b82 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":5},{"id":"d8178ae51d24afed3118573bcd3512cd01d8f8ef","author":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"date":"2020-04-27 16:31:01.000000000","message":"Patch Set 5:\n\nThank you everyone for the prompt reviews and the help provided to a new contributor! If these patches ever have issues, please do not hesitate to reach out.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c196cb0c30a2db68b8b1723766a4e34d7682236","revisions":{"3861255374719df241f4d99c8265fa7fb7bea517":{"kind":"REWORK","_number":1,"created":"2020-04-21 20:56:40.000000000","uploader":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"ref":"refs/changes/28/721728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/28/721728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/28/721728/1"}}},"commit":{"parents":[{"commit":"b51591b4f39b8a3feae89303bca61288e2d6ead6","subject":"Add support for release dates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/b51591b4f39b8a3feae89303bca61288e2d6ead6"}]}],"author":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 06:03:59.000000000","tz":-420},"committer":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 07:00:32.000000000","tz":-420},"subject":"Add cache support for release dates","message":"Add cache support for release dates\n\n- Also clean up use of internal data structures\n\nChange-Id: I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/3861255374719df241f4d99c8265fa7fb7bea517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/3861255374719df241f4d99c8265fa7fb7bea517"}]},"branch":"refs/heads/master"},"c862a74cf537a1beb4ff910db0128156c5fa1abd":{"kind":"REWORK","_number":2,"created":"2020-04-21 23:43:42.000000000","uploader":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"ref":"refs/changes/28/721728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/28/721728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/28/721728/2"}}},"commit":{"parents":[{"commit":"b51591b4f39b8a3feae89303bca61288e2d6ead6","subject":"Add support for release dates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/b51591b4f39b8a3feae89303bca61288e2d6ead6"}]}],"author":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 06:03:59.000000000","tz":-420},"committer":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 23:43:29.000000000","tz":-420},"subject":"Add cache support for release dates","message":"Add cache support for release dates\n\n- Also clean up use of internal data structures\n\nChange-Id: I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/c862a74cf537a1beb4ff910db0128156c5fa1abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/c862a74cf537a1beb4ff910db0128156c5fa1abd"}]},"branch":"refs/heads/master"},"79d30672c2f432f36410ab582ba60f71abec8ab9":{"kind":"REWORK","_number":3,"created":"2020-04-22 05:56:09.000000000","uploader":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"ref":"refs/changes/28/721728/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/28/721728/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/28/721728/3"}}},"commit":{"parents":[{"commit":"b51591b4f39b8a3feae89303bca61288e2d6ead6","subject":"Add support for release dates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/b51591b4f39b8a3feae89303bca61288e2d6ead6"}]}],"author":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 06:03:59.000000000","tz":-420},"committer":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-22 05:56:01.000000000","tz":-420},"subject":"Add cache support for release dates","message":"Add cache support for release dates\n\n- Also clean up use of internal data structures\n\nChange-Id: I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/79d30672c2f432f36410ab582ba60f71abec8ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/79d30672c2f432f36410ab582ba60f71abec8ab9"}]},"branch":"refs/heads/master"},"890219af751441937b4c1c304269872ced389384":{"kind":"REWORK","_number":4,"created":"2020-04-23 16:32:31.000000000","uploader":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"ref":"refs/changes/28/721728/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/28/721728/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/28/721728/4"}}},"commit":{"parents":[{"commit":"eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4"}]}],"author":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 06:03:59.000000000","tz":-420},"committer":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-23 16:32:25.000000000","tz":-420},"subject":"Add cache support for release dates","message":"Add cache support for release dates\n\n- Also clean up use of internal data structures\n\nChange-Id: I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/890219af751441937b4c1c304269872ced389384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/890219af751441937b4c1c304269872ced389384"}]},"branch":"refs/heads/master"},"6c196cb0c30a2db68b8b1723766a4e34d7682236":{"kind":"REWORK","_number":5,"created":"2020-04-23 17:13:54.000000000","uploader":{"_account_id":31872,"name":"Niraj Tolia","email":"ntolia@gmail.com","username":"ntolia"},"ref":"refs/changes/28/721728/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/28/721728/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/28/721728/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/28/721728/5"}}},"commit":{"parents":[{"commit":"eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4","subject":"Remove future imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/eac315ea0c2a4350ba378d746c2b8c0f1aeb34d4"}]}],"author":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-21 06:03:59.000000000","tz":-420},"committer":{"name":"Niraj Tolia","email":"ntolia@kasten.io","date":"2020-04-23 17:13:49.000000000","tz":-420},"subject":"Add cache support for release dates","message":"Add cache support for release dates\n\n- Also clean up use of internal data structures\n\nChange-Id: I45b9d5a0f3d63bf7bcfb39b3200442f68112bf97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/6c196cb0c30a2db68b8b1723766a4e34d7682236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/6c196cb0c30a2db68b8b1723766a4e34d7682236"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
