)]}'
{"id":"openstack%2Freno~733644","triplet_id":"openstack%2Freno~master~If90422fa95cb4eabc4c1722104cef3b28b7fca2d","project":"openstack/reno","branch":"master","topic":"add-encoding-config-option","hashtags":[],"change_id":"If90422fa95cb4eabc4c1722104cef3b28b7fca2d","subject":"Add config option to set default character encoding","status":"MERGED","created":"2020-06-04 15:27:01.000000000","updated":"2020-07-13 12:50:54.000000000","submitted":"2020-07-13 12:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733644-1594644548593-dfe663bb","meta_rev_id":"299594fcc65edd9e06b274054fa1c74369ef8f21","_number":733644,"virtual_id_number":733644,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-07-10 17:35:34.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 12:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 15:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-07-13 12:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-07-13 12:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 18:57:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-09 16:17:29.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-07-10 15:46:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-13 12:33:41.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-07-13 12:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c21e5bb672c9391dacb202e0aa98ed99e8e4befd","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-06-04 15:27:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5183d9cc0e4e3225f7c090e773dcd4fd41302749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:52:21.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b40c5d659c403a86d011cfa8d82245 : FAILURE in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b83b0ca135c44bb83e859576324de99 : FAILURE in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d2426806b284a61a158e6ad524d8edf : FAILURE in 8m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47bd9f78ec74d85976a4466282ceb41 : FAILURE in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7081c507244071b4c596f810d1abef : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef115fc84fa846ee8627ec6518780bbd : FAILURE in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"eb07ecfa8b99752272011b7fd1feb59de18abca7","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-06-04 21:10:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"264434d97fa12a9c2f141de9983e6de55abd5e54","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-06-04 21:11:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"69c840b34629f9a1f448891f89956dad4cf3ae7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 21:23:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f822ab5447497c88eec547b7b7ea64 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c3976c1415a4338a88b31ec7961fd6f : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ac341ceba8b4596be2cf436756957b1 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6650eed76b41d5b68363b0c6076263 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6db262b0074a23adcfca8f3c15ce38 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a67432e4b17144128339d78a714b738e : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c780442f25203796ce60adc4c150f4583acd1db0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 21:24:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b63cc595af6dfa13aaeb6aedd60a6deb711a9a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 18:57:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nThe handling in create.py is not consistent. What if a user creates his/her own template with specific encoding?","accounts_in_message":[],"_revision_number":2},{"id":"59142903c18afc492e0bbcdbf1dcf6172b8f397f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-05 18:58:57.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"017262f7e6e80ee03d6f564f4eac8775d0b9b4ba","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-06-05 19:19:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1893a510cb6d6b1d252bf4b6f5d1181e8b113509","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-06-05 19:20:14.000000000","message":"Patch Set 2:\n\n(3 comments)\n\noops, yeah good catch. I don\u0027t know how I missed that :)","accounts_in_message":[],"_revision_number":2},{"id":"ddc2fd88ad4dd90db1f025cadae8a9ae97b30409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:31:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/245c9ca5b4214d6c91e69c0d301e0d39 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec41eb9ba4349c4aac29124f5c1aa33 : FAILURE in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ffed40afd34c9787ececa68f5dbb38 : FAILURE in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22947716b9e444aab9c29f003e2fd6c4 : FAILURE in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4251f4ea9a7c4bc181f537cc883e693f : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06917750392f42f18cb4746bc2a07f21 : FAILURE in 7m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a73ac7822b1117fa1f1ea2590c3c14c29e11da40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-09 16:15:49.000000000","message":"Patch Set 3:\n\nAny updates on this?","accounts_in_message":[],"_revision_number":3},{"id":"96350c0ac641f56dae9aa2463a891ea0b25a4106","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-07-09 16:17:29.000000000","message":"Patch Set 3:\n\nmmm, is there a reason to have this, and not rely on utf8 by default?\n\nI am genuinely curious.","accounts_in_message":[],"_revision_number":3},{"id":"1020ebc4eed384bb8a3a630309329f2b28932077","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-07-09 18:14:31.000000000","message":"Patch Set 3:\n\nSorry I got sidetracked with other work and forgot to circle back and fix the tests here. I\u0027ll try to do it today or tomorrow.","accounts_in_message":[],"_revision_number":3},{"id":"0ce7d6d4ba77869c18c19a2b321e1e4f78f187e0","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-07-09 18:16:43.000000000","message":"Patch Set 3:\n\n\u003e mmm, is there a reason to have this, and not rely on utf8 by\n \u003e default?\n \u003e \n \u003e I am genuinely curious.\n\nI didn\u0027t think that was a safe assumption for all projects that could conceivably use reno. Like what if there was a git repo on windows only that was using the windows character encoding in their files and someone went to use it on linux? Like in the specific case where I I hit this issue the project already asserted all source files and release notes were utf8, but I didn\u0027t want to make having that assertion a hard requirement for using reno.","accounts_in_message":[],"_revision_number":3},{"id":"482bb449e4abaa55422eb0bf7daee762de9d1ed1","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-07-10 14:13:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9edc9785163adab973b7fcd92843967395882a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 14:27:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a162e606a9421ab3007d4e85da50f1 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383f9775a63a429e97ccc579e6fd6738 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8d71ce645c1404e845c1ad0d8e6b6c0 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2672d6e70f734b6289750a88c006d049 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c56678b3734d0d871e3cbcb5f4f41f : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50582fe731fa45d7a8e8c14c7cd19d56 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":4},{"id":"bbcd8308c856122dca63470db4a10c656d975161","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:46:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nLooks fine to me.","accounts_in_message":[],"_revision_number":4},{"id":"86a3376a9072f5db5f8654d9b46c4f08296173c1","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2020-07-10 17:35:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b457f277d1206ee31a2eac5760e554c8590cd024","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-07-13 12:33:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d58256f3f10449ba2c06c45497052de487f96541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:33:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f70d12046676819ba91607270e1069366c9b727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6eabd37e1bdcf186bedd20f5796983e5862543d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:49:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43dc4d7b8a394b588178b95663ed47c9 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/918382d52a9d4394853ee3f85ee11902 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/186b3e6d109f4c34b132f910ba023d2f : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476319a04ca84d918c51cc5af04e3d41 : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72181ba8837c408d892263fbd6d4046e : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":4},{"id":"b1a7262da2c6a183b3e767d65043799de6dca6d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:50:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761e4de720bd40d58a843c9de2649473 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"984bcba17e4e0b46763f42015d09680e5c5d5a04","revisions":{"35c6f7ee89b17965e7c5ca0e741e8b7e28210811":{"kind":"REWORK","_number":1,"created":"2020-06-04 15:27:01.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/44/733644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/44/733644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/44/733644/1"}}},"commit":{"parents":[{"commit":"1d503fa8ca0e44bbefdf5e5d744bf7645857c95d","subject":"Some cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d503fa8ca0e44bbefdf5e5d744bf7645857c95d"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 15:23:53.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 15:23:53.000000000","tz":-240},"subject":"Add config option to set default character encoding","message":"Add config option to set default character encoding\n\nThis commit adds a new config option \u0027encoding\u0027 which when set will take\na codec string from the stdlib codec library and use it as the encoding\nkwarg for all open() calls that are part of the reno commands. This will\nlet a project assert a specific encoding for all release note files,\nwhich can be important if the project is using special characters and\nhave contributors using multiple platforms or locales.\n\nChange-Id: If90422fa95cb4eabc4c1722104cef3b28b7fca2d\nStory: #2007757\nTask: #39959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c6f7ee89b17965e7c5ca0e741e8b7e28210811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c6f7ee89b17965e7c5ca0e741e8b7e28210811"}]},"branch":"refs/heads/master"},"f73804476b3f0fb60927a9f9715b9c5f98a885d0":{"kind":"REWORK","_number":2,"created":"2020-06-04 21:10:51.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/44/733644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/44/733644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/44/733644/2"}}},"commit":{"parents":[{"commit":"1d503fa8ca0e44bbefdf5e5d744bf7645857c95d","subject":"Some cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d503fa8ca0e44bbefdf5e5d744bf7645857c95d"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 15:23:53.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 21:10:44.000000000","tz":-240},"subject":"Add config option to set default character encoding","message":"Add config option to set default character encoding\n\nThis commit adds a new config option \u0027encoding\u0027 which when set will take\na codec string from the stdlib codec library and use it as the encoding\nkwarg for all open() calls that are part of the reno commands. This will\nlet a project assert a specific encoding for all release note files,\nwhich can be important if the project is using special characters and\nhave contributors using multiple platforms or locales.\n\nChange-Id: If90422fa95cb4eabc4c1722104cef3b28b7fca2d\nStory: #2007757\nTask: #39959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/f73804476b3f0fb60927a9f9715b9c5f98a885d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/f73804476b3f0fb60927a9f9715b9c5f98a885d0"}]},"branch":"refs/heads/master"},"1d21afe18b9d45a9b200c688999be0802c620425":{"kind":"REWORK","_number":3,"created":"2020-06-05 19:19:54.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/44/733644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/44/733644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/44/733644/3"}}},"commit":{"parents":[{"commit":"1d503fa8ca0e44bbefdf5e5d744bf7645857c95d","subject":"Some cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d503fa8ca0e44bbefdf5e5d744bf7645857c95d"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 15:23:53.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-05 19:19:49.000000000","tz":-240},"subject":"Add config option to set default character encoding","message":"Add config option to set default character encoding\n\nThis commit adds a new config option \u0027encoding\u0027 which when set will take\na codec string from the stdlib codec library and use it as the encoding\nkwarg for all open() calls that are part of the reno commands. This will\nlet a project assert a specific encoding for all release note files,\nwhich can be important if the project is using special characters and\nhave contributors using multiple platforms or locales.\n\nChange-Id: If90422fa95cb4eabc4c1722104cef3b28b7fca2d\nStory: #2007757\nTask: #39959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d21afe18b9d45a9b200c688999be0802c620425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d21afe18b9d45a9b200c688999be0802c620425"}]},"branch":"refs/heads/master"},"984bcba17e4e0b46763f42015d09680e5c5d5a04":{"kind":"REWORK","_number":4,"created":"2020-07-10 14:13:45.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/44/733644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/44/733644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/44/733644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/44/733644/4"}}},"commit":{"parents":[{"commit":"1d503fa8ca0e44bbefdf5e5d744bf7645857c95d","subject":"Some cleanups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/1d503fa8ca0e44bbefdf5e5d744bf7645857c95d"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-06-04 15:23:53.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2020-07-10 14:13:37.000000000","tz":-240},"subject":"Add config option to set default character encoding","message":"Add config option to set default character encoding\n\nThis commit adds a new config option \u0027encoding\u0027 which when set will take\na codec string from the stdlib codec library and use it as the encoding\nkwarg for all open() calls that are part of the reno commands. This will\nlet a project assert a specific encoding for all release note files,\nwhich can be important if the project is using special characters and\nhave contributors using multiple platforms or locales.\n\nChange-Id: If90422fa95cb4eabc4c1722104cef3b28b7fca2d\nStory: #2007757\nTask: #39959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/984bcba17e4e0b46763f42015d09680e5c5d5a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/984bcba17e4e0b46763f42015d09680e5c5d5a04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
