)]}'
{"id":"openstack%2Freno~772176","triplet_id":"openstack%2Freno~master~I30b5de3d5f35abb3b903e9ed9e4db42671421e88","project":"openstack/reno","branch":"master","topic":"restrict-subdir","hashtags":[],"change_id":"I30b5de3d5f35abb3b903e9ed9e4db42671421e88","subject":"Deny creating subdirectories","status":"MERGED","created":"2021-01-24 00:36:20.000000000","updated":"2021-02-22 20:52:17.000000000","submitted":"2021-02-22 20:51:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772176-restrict-subdir","meta_rev_id":"2941c0997eaaef08b365caea7414f48922665c67","_number":772176,"virtual_id_number":772176,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 20:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-22 20:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-22 10:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-22 20:41:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-24 01:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 10:19:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-02-22 20:41:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"ead0ffdd5522eb18acb0714e67c67d05e0e28d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 00:36:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3997fe84ca96c23022655afe2a08fb46faf7302b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 00:38:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c232a3f0bd42f3d4f1f9d39586f8086d6d6fc940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 01:20:10.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71093989680147c98e6336853b4406f4 : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0e01edee7e479882ff42e4d08c592f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc3e66155686466791cc61ac0b7c531a : SUCCESS in 7m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f494039698d4accb4616d1f70965f05 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e35999db3f24044801f7005e2595df0 : TIMED_OUT in 30m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f38a52035c60454e919fd9b8355c0acd : TIMED_OUT in 40m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f36734aec27b3dde6b5fada23fe20f45bb3eb25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 01:21:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"99f8185473f23a9c637cea9435bad1c6ec448a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 02:05:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c572cfa71d543eb92f24ced2d2d8be7 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a0faa3928940978a30f272bca1d60a : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d133653d761c46369f3d3defc7e8e660 : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208c8be7524a433b96154f103323fad9 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f44989e6a5248a4927c21c18f902885 : TIMED_OUT in 31m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e6cc0daf47b471d9d126e9e4d8ff5cd : TIMED_OUT in 40m 48s","accounts_in_message":[],"_revision_number":3},{"id":"6e6c9d0db059c63488ea268afc45af0cc86fb334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-19 13:53:19.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e30d179156ab82c7e146e9d8e07bf33cd30af235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:09:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/941565d1680f4b209edb0faa4ed636cd : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f511b50446242189ecdf09aa4f7f291 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e90de03da84a4297b248db4f65b190bd : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b97abc5ce99448e806f40b8efc44fca : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ca29946d7b48e7a5277a0e6d4eeaa1 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"71ccba65f4fb40368e6a764d9d3263c912960a7b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-22 10:19:26.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks good to me.\n\nNB: Depending on how likely that behavior was used in the past, this will require a major or minor version bump.","accounts_in_message":[],"_revision_number":4},{"id":"d7136f6bd1bad5d8f243ce732e67a4b96270d278","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-22 20:41:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aeddab18491736df65c76a521a654b09038e1c45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:42:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7ed71b903c824cd83564e40bbfd066012632d5cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:51:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fe61ddfbda4b6a9e9271fde0e14c1c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c033996470234143bbad1a3dd4900ba2 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521dae3fa18f4069a06a2fdbe4b69feb : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8dcf52580ab4945abc6b16628c33218 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"6dfc85bc03eac44b064755342c9b81a7662b76b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:51:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2941c0997eaaef08b365caea7414f48922665c67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 20:52:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af69424eb3bf4d439e53ddeed1f878e0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7dc390e6a6f2e43cbf299a3bb38c8e9c74f3e16a","revisions":{"88d1dca359e4a63efc03916c6be53023a912a3d1":{"kind":"REWORK","_number":1,"created":"2021-01-24 00:36:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/772176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/76/772176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/76/772176/1"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:36:18.000000000","tz":540},"subject":"Deny creating subdirectories","message":"Deny creating subdirectories\n\nCurrently if we put a path separator in slug, \"reno new\" command\ncreates subdirectories under the notes subdirectory. This usually\nresults in the situation where the note file is put to an unexpected\nplace.\nFor example we have seen several cases with reno files placed under\nreleasenote/note/release/note directory because the nodes subdirectory\npath is accidentally included in slug, like;\n\n$ reno new releasenotes/notes/something-new\nno configuration file in: ./releasenotes/config.yaml, ./reno.yaml\nCreated new notes file in releasenotes/notes/releasenotes/notes/something-new-2ff053c99583fc69.yaml\n\nThis change prohibits users from creating such subdirectries under\nthe notes subdirectory. The previous behavior can be restored by\nthe allow_subdirectories config option.\n\nChange-Id: I30b5de3d5f35abb3b903e9ed9e4db42671421e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/88d1dca359e4a63efc03916c6be53023a912a3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/88d1dca359e4a63efc03916c6be53023a912a3d1"}]},"branch":"refs/heads/master"},"bcb4e8916150dbd7b3e960dd3eb9b51edcbd50e8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-24 00:38:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/772176/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/76/772176/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/76/772176/2"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:37:59.000000000","tz":540},"subject":"Deny creating subdirectories","message":"Deny creating subdirectories\n\nCurrently if we put a path separator in slug, \"reno new\" command\ncreates subdirectories under the notes subdirectory. This usually\nresults in the situation where the note file is put to an unexpected\nplace.\nFor example we have seen several cases with reno files placed under\nreleasenote/note/release/note directory because the notes subdirectory\npath is accidentally included in slug, like;\n\n$ reno new releasenotes/notes/something-new\nno configuration file in: ./releasenotes/config.yaml, ./reno.yaml\nCreated new notes file in releasenotes/notes/releasenotes/notes/something-new-2ff053c99583fc69.yaml\n\nThis change prohibits users from creating such subdirectries under\nthe notes subdirectory. The previous behavior can be restored by\nthe allow_subdirectories config option.\n\nChange-Id: I30b5de3d5f35abb3b903e9ed9e4db42671421e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/bcb4e8916150dbd7b3e960dd3eb9b51edcbd50e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/bcb4e8916150dbd7b3e960dd3eb9b51edcbd50e8"}]},"branch":"refs/heads/master"},"262d27c1cf818659477ff14db73edcae0df9a33b":{"kind":"REWORK","_number":3,"created":"2021-01-24 01:21:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/772176/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/76/772176/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/76/772176/3"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 01:21:30.000000000","tz":540},"subject":"Deny creating subdirectories","message":"Deny creating subdirectories\n\nCurrently if we put a path separator in slug, \"reno new\" command\ncreates subdirectories under the notes subdirectory. This usually\nresults in the situation where the note file is put to an unexpected\nplace.\nFor example we have seen several cases with reno files placed under\nreleasenote/note/release/note directory because the notes subdirectory\npath is accidentally included in slug, like;\n\n$ reno new releasenotes/notes/something-new\nno configuration file in: ./releasenotes/config.yaml, ./reno.yaml\nCreated new notes file in releasenotes/notes/releasenotes/notes/something-new-2ff053c99583fc69.yaml\n\nThis change prohibits users from creating such subdirectries under\nthe notes subdirectory. The previous behavior can be restored by\nthe allow_subdirectories config option.\n\nChange-Id: I30b5de3d5f35abb3b903e9ed9e4db42671421e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/262d27c1cf818659477ff14db73edcae0df9a33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/262d27c1cf818659477ff14db73edcae0df9a33b"}]},"branch":"refs/heads/master"},"7dc390e6a6f2e43cbf299a3bb38c8e9c74f3e16a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-19 13:53:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/772176/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/76/772176/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/76/772176/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/76/772176/4"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 00:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-19 13:52:50.000000000","tz":540},"subject":"Deny creating subdirectories","message":"Deny creating subdirectories\n\nCurrently if we put a path separator in slug, \"reno new\" command\ncreates subdirectories under the notes subdirectory. This usually\nresults in the situation where the note file is put to an unexpected\nplace.\nFor example we have seen several cases with reno files placed under\nreleasenote/notes/releasenote/notes directory because the notes\nsubdirectory path is accidentally included in slug, like;\n\n$ reno new releasenotes/notes/something-new\nno configuration file in: ./releasenotes/config.yaml, ./reno.yaml\nCreated new notes file in releasenotes/notes/releasenotes/notes/something-new-2ff053c99583fc69.yaml\n\nThis change prohibits users from creating such subdirectries under\nthe notes subdirectory. The previous behavior can be restored by\nthe allow_subdirectories config option.\n\nChange-Id: I30b5de3d5f35abb3b903e9ed9e4db42671421e88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/7dc390e6a6f2e43cbf299a3bb38c8e9c74f3e16a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/7dc390e6a6f2e43cbf299a3bb38c8e9c74f3e16a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
