)]}'
{"id":"openstack%2Freno~774867","triplet_id":"openstack%2Freno~master~I77ae7c2d5716430bb0a83d4f1ff82c75a8b695d6","project":"openstack/reno","branch":"master","topic":"fix-tag-selection","hashtags":[],"change_id":"I77ae7c2d5716430bb0a83d4f1ff82c75a8b695d6","subject":"fix: order tags to be sure to favor release tag over pre-release tags","status":"MERGED","created":"2021-02-10 14:19:40.000000000","updated":"2021-02-18 07:00:12.000000000","submitted":"2021-02-18 06:58:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774867-fix-tag-selection","meta_rev_id":"ff770a4de1d92009dc7e298731b62159e065dd8f","_number":774867,"virtual_id_number":774867,"owner":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 06:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2021-02-10 16:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2021-02-18 06:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 14:27:44.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2021-02-10 15:13:32.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-02-10 15:29:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 14:15:21.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-02-16 09:40:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2021-02-18 06:44:48.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"d5993db3823e0b4607ab4a4d83cfb6060700e9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-10 14:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a16f4865397ddf8790a11e258f38cc713e5faba3","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-02-10 14:27:44.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThe code changes look good. I have one question about the comment that\u0027s included, though.","accounts_in_message":[],"_revision_number":1},{"id":"4ff634f1cc3251d5cec86dba9683fce1ab623571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-10 14:42:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"205a12c895dd8f160037f5dc9c8d6bfe42e283f8","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-10 14:43:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"280ba87b066fccb02c10ef9f4de82fc2894c3d5b","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-02-10 15:13:32.000000000","message":"Patch Set 2: Code-Review+2\n\nThis looks good. It might have been safer to sort the values at the point where the ordering is important, but having them always be sorted isn\u0027t wrong.\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"e9062c5505e7c68b4e51cd78463a6a5a37da4b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 15:29:04.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf60cd6797646bcb9a692162cc72a45 : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b1fed99f0f042259945b8eaa4c0f818 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5e5a66cfdc348d8aaae474df0cb8965 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0731f01f8f544a8cb17e74ae8b7b1de5 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722fb5bc68014e9b8b00a8075cd44517 : TIMED_OUT in 30m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3376c0351a0a492c8ce8cccba534ce3f : TIMED_OUT in 40m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7c2934715686a72c68ddec23a5f9f916ee6d7b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-10 16:37:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6d97b6033009a409d1b3ccd88a698dd1987f7ed","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-10 16:37:54.000000000","message":"Description of patch set 3 set to \"Fixed pep8\"","accounts_in_message":[],"_revision_number":3},{"id":"399faa3aafb867bb45884b0c523e908d2ef63576","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-02-10 16:57:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5efe4408d1957b5cd51a529f1c6d47acad475498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:25:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a645f9f26ae462aad905d1e814ebc20 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f63a4f555ed24f4fb608840472d3b3b1 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37c78477e64a4edca1c2091b16081393 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9054993b66a45b6873152767ad3a758 : POST_FAILURE in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2df43c0e0d4886b57fd8475fc8f61f : TIMED_OUT in 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86e82bca12c64bc1a55f6e3fa7883f17 : TIMED_OUT in 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e58ef73ad696d67b2c79b61bbd2805d01630f31e","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-11 08:56:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef864cadb045d75175a3f69038139bba7d95dea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 09:58:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da34163bcfa34e9abcc16dd5e9df120b : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a246f56b64244dc9a10908662160faa : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13c9ba2a59df43bb925221fdbe0d532a : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3dd243a5b4c42b48ab7bf6aacfd635a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62a9416109c4e84b147822ca361cf01 : TIMED_OUT in 31m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14eb91ad330f41549363b9e7ec910430 : TIMED_OUT in 41m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9c5f6745c7b235edee3b6e2f2da359a353c72f06","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-11 12:31:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2ce7842611696f74cff750b8daaac67e7fdcfee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:39:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d468ba83eee43cf961a4079f620335b : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14545ab4e0334afebe2201e140e198fe : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adbc8fb7beec49e79ff4c66cc681d4d9 : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c861dc613241afa065fe470242c4e8 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfbf56cac4574390b39b962917dd5725 : TIMED_OUT in 31m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12309c5fd0104e4dbaec0a35bdc4fa86 : TIMED_OUT in 41m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8713e884508eb5a2285d718654951748bdcc724b","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-11 13:55:25.000000000","message":"Patch Set 3:\n\nDoug, should I keep rechecking or is there any ongoing known issue?","accounts_in_message":[],"_revision_number":3},{"id":"a3f79ea435d4313c2db5a55771a9c3dd21f5a255","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-02-11 14:15:21.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Doug, should I keep rechecking or is there any ongoing known issue?\n\nI don\u0027t really work here any more, so I\u0027m not sure. :-)\n\nHervé do you know?","accounts_in_message":[],"_revision_number":3},{"id":"3ae2f334bd0811a5351121321553f7887da9e871","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-12 17:09:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4f36ec8786fecfac3543f093f6ceb551f96d7391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 17:55:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb8d6718b2b4324bb92d293ccf54dc9 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b34cff841d0745c9829c982ab50a0315 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f72c1d149ab4b638f48042bd74f8d7e : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb8c23e22ec48379f4cb129172751ea : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c19905d6e943429a9b8c02fa73ed7372 : TIMED_OUT in 31m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a2ba890fe024aaf9321547625df5335 : TIMED_OUT in 41m 16s","accounts_in_message":[],"_revision_number":3},{"id":"673d54dc3e82e6fbf5abe5608f85ffb1d5f13831","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-14 08:50:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9f6c849d8aed4af36663584bfb65340e86f1250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 09:35:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42cca13791f241fdbab4da3db93db2ad : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fca7974bd8346d784242808f80e2a02 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de48b84a430f4c72817f02e4f1acc768 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e0ab7d7e014274aaaa363640ec6e2f : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95c5fab91f24bbb9924aac0705aec0d : TIMED_OUT in 30m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3247c7a81eb433aab7cf50678614ee4 : TIMED_OUT in 40m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8a0a0d16abe7ec12eb6eadfbfc9566d612b3f42a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-02-14 15:49:21.000000000","message":"Patch Set 3:\n\nIt looks like the issue is pip\u0027s new resolver taking a long time to made a decision about the dependencies in the docs job. That job is using the lower-constraints.txt, so I wonder if we can improve performance by being more specific there?\n\nhttps://zuul.opendev.org/t/openstack/build/e95c5fab91f24bbb9924aac0705aec0d/log/tox/docs-1.log#144","accounts_in_message":[],"_revision_number":3},{"id":"27cd8c774dd4a4471f35fd44d54f04e74d3eced0","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-15 08:41:13.000000000","message":"Patch Set 3:\n\nI\u0027m not sure what that means. How be more specific, I see lower constraints as pinning versions, no? 😞","accounts_in_message":[],"_revision_number":3},{"id":"94d9f2ad7449e8d2522001d67eb7bd1053652c9f","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2021-02-15 08:44:33.000000000","message":"Patch Set 3:\n\nI just saw your others reviews 😊 👍","accounts_in_message":[],"_revision_number":3},{"id":"1b43ef4ad8d4a6b7088a081d03ae19711bcd6254","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-02-16 09:40:44.000000000","message":"Patch Set 3:\n\nrecheck - The lower-constraints job was dropped, this should pass tests now.","accounts_in_message":[],"_revision_number":3},{"id":"4506bff7336d0563263fafca81cc2fb4b19017b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 09:55:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554c6b6aa67046ea9625a7b9af6feff4 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4db477960a3d439bb7dd5721d1c7b689 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4811753d70d43548343454d7508f0a7 : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18c7f9ea2bb445048c22944723e7b873 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c3fefd634c4d2980725168c476785a : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"2c8fcd7f20a7dda5e57b9250923dd464753ee062","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-18 06:44:48.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e356a6536740097e1fdc9d3e40abb99c44aeb7b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 06:45:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b86a5a7432ec20c4ed3bda423195df057350dc07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 06:58:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973e460c751a4ca7ba4e3fc92d0cae7d : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d595d77512d649b5add98fb92ef613f2 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de65938768bd4078b6c68f3c1ecd6ed9 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e6c9d7277b4b7e94dfad22e3be3639 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"778ffa3ced426ff3d75c09d58d32f6282f3eef51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 06:58:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ff770a4de1d92009dc7e298731b62159e065dd8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 07:00:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677ab9d9609e4c6ba7b08fa443862a32 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf6cbb73b233abbbff9066e8a806f762e0073d0d","revisions":{"9291ac43a670e191f9d3a2beda24c1a69c297025":{"kind":"REWORK","_number":1,"created":"2021-02-10 14:19:40.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/67/774867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/67/774867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/67/774867/1"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 14:18:18.000000000","tz":60},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 14:18:18.000000000","tz":60},"subject":"fix: order tags to be sure to favor release tag over pre-release tags","message":"fix: order tags to be sure to favor release tag over pre-release tags\n\nWhen a commit matches multiple tags, it\u0027s usually preferable to select the\nrelease tag. The typical use case is tagging a release candidate and then\ntagging it with the release tag with no new commits.\n\nChange-Id: I77ae7c2d5716430bb0a83d4f1ff82c75a8b695d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9291ac43a670e191f9d3a2beda24c1a69c297025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9291ac43a670e191f9d3a2beda24c1a69c297025"}]},"branch":"refs/heads/master"},"8a3f4aa32254cf3fe78e0571625e26f9640408a5":{"kind":"REWORK","_number":2,"created":"2021-02-10 14:42:22.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/67/774867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/67/774867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/67/774867/2"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 14:18:18.000000000","tz":60},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 14:42:16.000000000","tz":60},"subject":"fix: order tags to be sure to favor release tag over pre-release tags","message":"fix: order tags to be sure to favor release tag over pre-release tags\n\nWhen a commit matches multiple tags, it\u0027s usually preferable to select the\nrelease tag. The typical use case is tagging a release candidate and then\ntagging it with the release tag with no new commits.\n\nChange-Id: I77ae7c2d5716430bb0a83d4f1ff82c75a8b695d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/8a3f4aa32254cf3fe78e0571625e26f9640408a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/8a3f4aa32254cf3fe78e0571625e26f9640408a5"}]},"branch":"refs/heads/master"},"cf6cbb73b233abbbff9066e8a806f762e0073d0d":{"kind":"REWORK","_number":3,"created":"2021-02-10 16:37:02.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/67/774867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/67/774867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/67/774867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/67/774867/3"}}},"commit":{"parents":[{"commit":"35c0a43c390f66e6785e78ad4e0f71561e9f052a","subject":"raise the minimum version of pyyaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/35c0a43c390f66e6785e78ad4e0f71561e9f052a"}]}],"author":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 14:18:18.000000000","tz":60},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2021-02-10 16:36:49.000000000","tz":60},"subject":"fix: order tags to be sure to favor release tag over pre-release tags","message":"fix: order tags to be sure to favor release tag over pre-release tags\n\nWhen a commit matches multiple tags, it\u0027s usually preferable to select the\nrelease tag. The typical use case is tagging a release candidate and then\ntagging it with the release tag with no new commits.\n\nChange-Id: I77ae7c2d5716430bb0a83d4f1ff82c75a8b695d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/cf6cbb73b233abbbff9066e8a806f762e0073d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/cf6cbb73b233abbbff9066e8a806f762e0073d0d"}]},"branch":"refs/heads/master","description":"Fixed pep8"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
