)]}'
{"id":"openstack%2Freno~787277","triplet_id":"openstack%2Freno~master~I08f258206290fdc1dcb9fcfc078e30384667c925","project":"openstack/reno","branch":"master","topic":"configurable-default-branch","hashtags":[],"change_id":"I08f258206290fdc1dcb9fcfc078e30384667c925","subject":"Handle repositories with a default branch other than master","status":"MERGED","created":"2021-04-20 18:28:48.000000000","updated":"2021-04-22 09:51:18.000000000","submitted":"2021-04-22 09:49:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787277-configurable-default-branch","meta_rev_id":"b477f1f32f790127fa81c3d689f2ee7dafc4e368","_number":787277,"virtual_id_number":787277,"owner":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 09:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-04-22 09:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-04-20 18:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-04-22 09:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 18:34:00.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-04-20 18:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 09:20:14.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"40c06a652ceb1710901872e49ae5984e6ea0523d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2021-04-20 18:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eded542f7348505ed757f1f0fe49f04feca68079","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-04-20 18:34:00.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks good. It would be nice to have a test for it, but if the regular tests work that\u0027s probably good enough.","accounts_in_message":[],"_revision_number":1},{"id":"13ba55d0c513d9e233fb70065e766ec930dd440a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c51c55e2804a57977828198937f988 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb78ccff09aa47ff91b0c3190680fceb : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/302c6cd5bed242c3b7a50865f63424eb : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e85dbf3d1d6b4e7dbe9cd9c3da02da62 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4113cf93227c4622abab81f2dbbea074 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"467ba05d7250340880d8d3956ab5b682b711372b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-04-22 09:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d894c85924501c17d37ac6b2e7ec6704eb301f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:20:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27ef86e6d67fe5a27a91590622c4677e9c918914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:49:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e205e70f5f4342238068cb21832b6857 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45b64e4290364ff89094b8fa1e6a484e : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c53debdbdad4b36ba148d77f4688ede : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6522d6a7ee814f22b81dc7a6661e9d7e : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2b89c2c81680394037b8eb29e5e99d9592dedea3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:49:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b477f1f32f790127fa81c3d689f2ee7dafc4e368","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db83fb19e5c43499dfb9c963b3b84ee : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed6bbae82e01edf781534d9e9cce21e0c55a49a6","revisions":{"ed6bbae82e01edf781534d9e9cce21e0c55a49a6":{"kind":"REWORK","_number":1,"created":"2021-04-20 18:28:48.000000000","uploader":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"ref":"refs/changes/77/787277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/77/787277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/77/787277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/77/787277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/77/787277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/77/787277/1"}}},"commit":{"parents":[{"commit":"e45092be853ae0366aa2ab4ae049cceaf2932bd9","subject":"Merge \"Deny creating subdirectories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e45092be853ae0366aa2ab4ae049cceaf2932bd9"}]}],"author":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2021-04-20 18:26:18.000000000","tz":-240},"committer":{"name":"Matthew Treinish","email":"mtreinish@kortar.org","date":"2021-04-20 18:26:18.000000000","tz":-240},"subject":"Handle repositories with a default branch other than master","message":"Handle repositories with a default branch other than master\n\nThis commit adds a new configuration option `default_branch`, which is\nused to specify the default/base branch a repository is using.\nPreviously reno assumed the default branch was always named \u0027master\u0027\nwhich causes commands such as \u0027reno lint\u0027 to fail when it is run on a\nrepository that uses a different default branch name such as \u0027main\u0027.\nThis config option gives users an option to specify which branch should\nbe used for cases like that.\n\nChange-Id: I08f258206290fdc1dcb9fcfc078e30384667c925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ed6bbae82e01edf781534d9e9cce21e0c55a49a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ed6bbae82e01edf781534d9e9cce21e0c55a49a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
