)]}'
{"id":"openstack%2Freno~800205","triplet_id":"openstack%2Freno~master~I43d98cc75332677c604e53a5b7a21239505ab4a4","project":"openstack/reno","branch":"master","topic":"story/1670421","hashtags":[],"change_id":"I43d98cc75332677c604e53a5b7a21239505ab4a4","subject":"loader: Validate section keys","status":"MERGED","created":"2021-07-09 11:07:30.000000000","updated":"2021-07-19 13:19:44.000000000","submitted":"2021-07-19 13:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"800205-story/1670421","meta_rev_id":"eaf5dad92bf31eace42434ae6d5c2f11f82dae25","_number":800205,"virtual_id_number":800205,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-19 13:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-07-19 13:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-07-13 21:20:24.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2021-07-19 13:00:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 11:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 12:32:12.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2021-07-13 21:20:24.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-07-19 13:00:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"fd34191dff4170ac7057263384680d09e37109b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-09 11:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d63fe3b34b927cc3391b667df2d252ecaeec4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 11:20:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79732a8732849c2b02d01cceb9e31f2 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d9bde79517b4c5680f9bf8ad73adf91 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bf356b6645f4c499b70f85e750dd738 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1596f1d5ecd4272ad9a7c2b2ab68e8d : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed6ab59f1f647048b5aec09a5c3f1ae : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3e18b20a1b7a6e4fac2eded708f83ce6a6eb211e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-07-09 12:32:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f979cfdd6681dc59e3bdda9033c3c8f861e9aaf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 14:26:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be4833d6d07dd85c2fb62a5ddc16fe98998b84d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-12 14:27:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e550ff9a1c98051d64de841e3c0b423b7e342d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 14:37:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116a44e48fa1440faf71fc95c54dc335 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb8778eb286849f5b32d57e7da1e5ef0 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ff22450252b4460b665b984866d0947 : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0de1d42d5b6461c9f41c46b46496488 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6df30c74b4343b4816fe64bb73bbe24 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a4d364626053b8502ddba5a99f3d5ffaf8db7681","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-07-13 21:20:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9a658a88daabc881e9dd096298ce7d194b5e4228","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-07-19 13:00:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d22e9335ab15e405218e6d1a3eda6129041b19af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:00:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee5d4e6c6623ba77f397f3ac2eda53ddae8b725d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:15:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527d5765f00f4bb8919ad3d9023e7eb6 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05fb46f1b494482383b4adce9b8b2e9f : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4303b27618bd4383936c685bfc5f061a : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ce753712c448ccbed60109897c903d : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"285056ac90db7d3ad8ec9d81c4cb9ae5b5d39c25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"eaf5dad92bf31eace42434ae6d5c2f11f82dae25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 13:19:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273136cb962942dab5be7f4e9138d888 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a80087d923d3fb34c3db6d2b70e317716cd92e1","revisions":{"e9525543935b2d572c99eb056826f99e3de524c4":{"kind":"REWORK","_number":1,"created":"2021-07-09 11:07:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/800205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/05/800205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/05/800205/1"}}},"commit":{"parents":[{"commit":"681ddf5bd040e7672e6408d322994803a767624f","subject":"loader: Improve tests and logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/681ddf5bd040e7672e6408d322994803a767624f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-09 11:01:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-09 11:04:58.000000000","tz":60},"subject":"loader: Validate section keys","message":"loader: Validate section keys\n\nThis stops silly typos making their way into release notes.\n\nChange-Id: I43d98cc75332677c604e53a5b7a21239505ab4a4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nStory: 1670421\nTask: 6737\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e9525543935b2d572c99eb056826f99e3de524c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/e9525543935b2d572c99eb056826f99e3de524c4"}]},"branch":"refs/heads/master"},"8a80087d923d3fb34c3db6d2b70e317716cd92e1":{"kind":"REWORK","_number":2,"created":"2021-07-12 14:27:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/800205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/05/800205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/05/800205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/05/800205/2"}}},"commit":{"parents":[{"commit":"c4b517292e7e457901ace60471465ce093f0f5a1","subject":"loader: Improve tests and logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/c4b517292e7e457901ace60471465ce093f0f5a1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-09 11:01:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-12 14:27:19.000000000","tz":60},"subject":"loader: Validate section keys","message":"loader: Validate section keys\n\nThis stops silly typos making their way into release notes.\n\nChange-Id: I43d98cc75332677c604e53a5b7a21239505ab4a4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nStory: 1670421\nTask: 6737\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/8a80087d923d3fb34c3db6d2b70e317716cd92e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/8a80087d923d3fb34c3db6d2b70e317716cd92e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
