)]}'
{"id":"openstack%2Freno~850151","triplet_id":"openstack%2Freno~master~Id26e954f347c35703e506abdc8ecf5146f2efb61","project":"openstack/reno","branch":"master","topic":"story/2010155","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-21 08:46:42.000000000","reason":"Change was submitted"},"2472":{"account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"last_update":"2022-07-19 13:52:14.000000000","reason":"\u003cGERRIT_ACCOUNT_2472\u003e replied on the change","reason_account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}}},"hashtags":[],"change_id":"Id26e954f347c35703e506abdc8ecf5146f2efb61","subject":"Don\u0027t attempt to pull from empty cache file","status":"MERGED","created":"2022-07-18 11:52:53.000000000","updated":"2022-07-21 08:47:48.000000000","submitted":"2022-07-21 08:46:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850151-story/2010155","meta_rev_id":"a4674473e590e854ac4a472cf286abd9a2f7c8fb","_number":850151,"virtual_id_number":850151,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-21 08:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-07-21 08:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-07-19 13:52:14.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2022-07-21 08:34:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 12:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 20:53:31.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2022-07-19 13:52:14.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2022-07-21 08:34:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"a01a259941ad7bd8cbe9709637b93f26d1218e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-18 11:52:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e03ffcd889b2bb39ababaae1de0a54e6158504f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 12:08:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb46c78adc1348d3af71b20ebcd0822b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3195ceaf12874416a7875252705f3d3d : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23be1cc01b9648529fc36917ffc372a7 : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d9a7ba565fb42ab9d0a8f6576e04a60 : SUCCESS in 9m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2a32c1558cb4bff98c3904d315d8826 : RETRY_LIMIT in 2m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37e4c29429394a50922ab99f92863ae1 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac8755d30064d96ba872fa27221c082 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cd000a882905c87d424d00bc6225719b19d89d35","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-07-18 20:53:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b674ab8e4250e61f37c732836e0fb51aa35afd0f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-19 11:02:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46c16fcd745894472dbf31e5742dfbc770627387","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-07-19 12:45:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dba6241105d71339a618ae97cd723f899232297c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-19 12:48:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9621b2547a219543867ead431f7df2c2bb668cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-19 12:50:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8eeccff2b6571dd80285f94e202d2f095c809e06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-19 12:50:38.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"80e9ef2e783711a5d812322b83f2bc7ec823a13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:10:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6ff3e5b0f14cb388fff16aed064df9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d18888b5ff14eefa5d71895bc47ec46 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38699e6dd0ed49d181da0bd80ee469f9 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3960537010b84ba98e49e5b0f59c4978 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/248cba4a5991459883066e778676b0b3 : RETRY_LIMIT in 3m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c521c455e437449896213c205ee8d1eb : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e398267dc20442029b73b9da13951f18 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"426fdd2ebb51c5ec1a55ad8e64b18f4015a3a0fd","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-07-19 13:52:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"167dc35963a3a52154f5e0a1e3fa55b54f2d250e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-07-21 08:34:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d470db68bdc4adb8fcce1e85499c9e489e2cf63d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:35:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d91f4786071eab082f796af919246c675e470ac3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:46:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d47bdd424b941c69186a3262d00d0fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b3e388ee1c4037b8ee61a39fb6af72 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d40d6010ba5a4f17a73d196eba9c6038 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5637544655304272adb6e1f497ebfefb : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4127bc08857748cdb62e651a05e9f8f5 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"35e0592a8a13544e508368b402e79610da48d4c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:46:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a4674473e590e854ac4a472cf286abd9a2f7c8fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:47:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b70903fc3d54e29bb09c59b0395b7c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ffa90cb3874872b113fa80aa57d05f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aae5632aa19242bdd5e5abc2d919cdcb0b1b9e15","revisions":{"575b698469e597017a8fccfb2c09b99a629669e7":{"kind":"REWORK","_number":1,"created":"2022-07-18 11:52:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/850151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/51/850151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/51/850151/1"}}},"commit":{"parents":[{"commit":"9700000b3540fca69595b29e3b77df85e8d6da4f","subject":"Update CI to use unversioned jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9700000b3540fca69595b29e3b77df85e8d6da4f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-18 11:42:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-18 11:42:43.000000000","tz":60},"subject":"Don\u0027t attempt to pull from empty cache file","message":"Don\u0027t attempt to pull from empty cache file\n\nChange-Id: Id26e954f347c35703e506abdc8ecf5146f2efb61\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: 2010155\nTask: 45808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/575b698469e597017a8fccfb2c09b99a629669e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/575b698469e597017a8fccfb2c09b99a629669e7"}]},"branch":"refs/heads/master"},"aae5632aa19242bdd5e5abc2d919cdcb0b1b9e15":{"kind":"REWORK","_number":2,"created":"2022-07-19 12:50:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/850151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/51/850151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/51/850151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/51/850151/2"}}},"commit":{"parents":[{"commit":"9700000b3540fca69595b29e3b77df85e8d6da4f","subject":"Update CI to use unversioned jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9700000b3540fca69595b29e3b77df85e8d6da4f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-19 12:50:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-19 12:50:00.000000000","tz":60},"subject":"Don\u0027t attempt to pull from empty cache file","message":"Don\u0027t attempt to pull from empty cache file\n\nChange-Id: Id26e954f347c35703e506abdc8ecf5146f2efb61\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStory: 2010155\nTask: 45808\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/aae5632aa19242bdd5e5abc2d919cdcb0b1b9e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/aae5632aa19242bdd5e5abc2d919cdcb0b1b9e15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[]}
