)]}'
{"id":"openstack%2Freno~863846","triplet_id":"openstack%2Freno~master~I3abd71b9b18e627b3185b7a07a7a33f516ec3a72","project":"openstack/reno","branch":"master","topic":"drop-python-dev-from-bindep","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-07 14:07:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3abd71b9b18e627b3185b7a07a7a33f516ec3a72","subject":"Remove python-dev from bindep","status":"MERGED","created":"2022-11-07 10:05:16.000000000","updated":"2022-11-07 14:08:46.000000000","submitted":"2022-11-07 14:07:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863846-drop-python-dev-from-bindep","meta_rev_id":"a93a2c12c9aef97f29376236611e2ac303e9aad6","_number":863846,"virtual_id_number":863846,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 14:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-11-07 13:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-11-07 13:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2022-11-07 13:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 10:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-07 13:55:04.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2022-11-07 13:57:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"d82a324a8c96e0cd20d7b3e6640037d70da96403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-07 10:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813a97c6dabf3095cd7733604194e98973167c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:12:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228c166c56ad4397b0acbf4319e72111\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982fd6720ec742cca5ebaf123526fa87 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/850907b08cf645fbbaa0bb8f30e7f3bc : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6d36a66126e468cb0c59a7102cb8896 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e2d34aa4bd4c61afb7350b01f00ce7 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7916d7fd734c43bd506ead20758551 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7e370c66a5140e12d256a96b00e6c5e0b86c21e3","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-11-07 13:55:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60a9eb133e4f9d7b924a4cb58ef9beff397d82f0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-11-07 13:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8507513c9f67f860bfb4de812e4790f43647a97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 13:58:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de9c4d77de8ae4668f3284857f09d6f36d101c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:07:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840dbf9959b44a0dac00837a18bb30eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3644b992b6104a5da8a2495456146a27 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39f0ca2175d047a9bfd0b53f309b1b40 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae25bd7f6570481caf9003ac13707652 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37074e2dcaaa43f784aa2c7f12304809 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"16a6f87013370d6074ca98bf3d93443f8c0853b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:07:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a93a2c12c9aef97f29376236611e2ac303e9aad6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2207fc6d787d4b75b65c93bfb783d80e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff919e2496c747b8bd97a03ded979f39 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04233e0eae9d688a3eb0630443a30b860c04601e","revisions":{"04233e0eae9d688a3eb0630443a30b860c04601e":{"kind":"REWORK","_number":1,"created":"2022-11-07 10:05:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/863846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/46/863846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/46/863846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/46/863846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/46/863846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/46/863846/1"}}},"commit":{"parents":[{"commit":"5c31cd7c0aa2a4dbdfa71eda73cf5f66c23193ec","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/5c31cd7c0aa2a4dbdfa71eda73cf5f66c23193ec"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:05:12.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-07 10:05:12.000000000","tz":60},"subject":"Remove python-dev from bindep","message":"Remove python-dev from bindep\n\nIt is no longer supported by jammy and lead us to the following errors with the announce-release job.\n\n```\nNo package matching \u0027python-dev\u0027 is available\n```\n\nChange-Id: I3abd71b9b18e627b3185b7a07a7a33f516ec3a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/04233e0eae9d688a3eb0630443a30b860c04601e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/04233e0eae9d688a3eb0630443a30b860c04601e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[]}
