)]}'
{"id":"openstack%2Freno~873165","triplet_id":"openstack%2Freno~master~Ib7b4781de81e19d11a7acb863204067c0946827b","project":"openstack/reno","branch":"master","topic":"nested-headers","attention_set":{},"removed_from_attention_set":{"5196":{"account":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"last_update":"2023-03-30 12:03:32.000000000","reason":"Change was submitted"},"35758":{"account":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"last_update":"2023-03-30 12:03:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7b4781de81e19d11a7acb863204067c0946827b","subject":"Add support for subsections","status":"MERGED","created":"2023-02-08 17:55:19.000000000","updated":"2023-03-30 12:04:29.000000000","submitted":"2023-03-30 12:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"873165-nested-headers","meta_rev_id":"9593802b096363e8e01dca8a3468e817a135cd13","_number":873165,"virtual_id_number":873165,"owner":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-30 12:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-03-23 14:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-03-30 11:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-03-30 11:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-08 18:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 14:58:14.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"CC"},{"updated":"2023-03-07 17:18:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-03-23 14:32:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-03-30 11:53:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"32f88d4fd94c5b73bd96917385284f5b700d72bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-02-08 17:55:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"390ee2d07da8887a51a4719319cbaa7641555eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 18:19:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b299361f2d4272b07104d4247fd24d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06280fb4c4945d2bbe8725e005b8eac : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73333f06ceec40f7b63e6c2ffe75594d : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b41170ae2a994010909445618a06d8ba : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e3e93e3b384ff48c65588cdbeac5ea : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84b5000e950441c80c927c9e709258d : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"805c18c99202c07506dfb1d03a1d6d5d352d3335","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2023-02-09 14:58:14.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8eb69ca13d9e41f21d19228a9b472b315bcdd95","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-02-10 02:59:23.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6ccabfbba140764a40c72a5ce3700238714ce71c","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2023-02-10 13:04:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e601e078e047fe66497638418bb96c4b37493397","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-02-10 19:29:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40477d0818abeafbebed5f12d36f0657b777fff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-02-10 20:32:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ac7fbf2fdf60ba043bde9e938a79017db1bf97f","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-02-10 20:35:33.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e3e3029a9beef44b7b8ab0cbac3def8db3524c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:40:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674cfa6e861b48f1a02ae4382bd3933c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05227eee1c4b405ca4d1199f841f7aed : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acf0445551de4794a60233d75782bc3e : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c9e99e014bc4b4dae74b7f254a6cb29 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e714cd8b9d24ba4913d264a6f5d9de9 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a0504e6ce3431e8dc7a10c40d12af6 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"beddf8f9613bc7de6cb605c9ac9e6851043ebc00","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-07 17:18:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6917460fbc66a7a63762fde6ee12488df724abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 19:57:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a48d332ac81367277fe15e5bbf90dc2760ad8f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 20:05:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8883a1484d414abc8515c010205aa359\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2c41f279a443199a1f58a4f3cde1b6 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb4c974ed6454747a6ecba1b4952c829 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88ffd3f98a8d40a1a7ae3e355f5298da : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96104d80c3b840cd93e8960708982b8b : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c8c4a662bd40fbb92e29eb648454b4 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"d55353329872183503f1058ae71eb24d61ebdd09","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-03-23 14:32:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e1110a19a16defe22c749806e5f7682f6896706e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-30 11:53:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"793b9f6d2705a7f731b495acbffb12d45e7728f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:53:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"adeea265d994a91e9371b61515c05a1551d1cd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:03:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25689972d7a34bdc8941cde41f1448cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85fdadb266540ef9c2c1143a8d6f09a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d6521304064561911dfe01be0febe1 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74b72aeeb2bf4478a1c24292c912f024 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d021b268244b4da3611cecc833c896 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"dcd8b71dc8dccc2ff4b79267cb38318126660752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:03:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9593802b096363e8e01dca8a3468e817a135cd13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:04:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a8620cc4c645a986754697ba732238\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae37813a4f040bbaf727323d50fbfda : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee73c9409e4bbceec565a42c7090a72455b1dcfc","revisions":{"9f641d2efaa402bedacc0bf40d003041b1a6b2b8":{"kind":"REWORK","_number":1,"created":"2023-02-08 17:55:19.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/65/873165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/65/873165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/65/873165/1"}}},"commit":{"parents":[{"commit":"919210c386a970ea0b28e4e7df3bb783c9f3895c","subject":"Refactor Section config to use NamedTuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/919210c386a970ea0b28e4e7df3bb783c9f3895c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-02-07 22:59:28.000000000","tz":-420},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-02-08 17:54:47.000000000","tz":-420},"subject":"Add support for subsections","message":"Add support for subsections\n\nCloses https://storyboard.openstack.org/#!/story/2010375. This\nfeature is useful for longer changelogs to add more order.\n\nSubsections will appear underneath their top-level section, in\nthe order defined in the config file. They will use a ^ underline\nrather than -.\n\nUsers define subsections as a list of 2-pair IDs and titles,\nat the end of the top-level section\u0027s entry. This syntax ensures\nthat we correctly link the subsection to its parent section because\nthey all appear in the same entry.\n\nLike normal sections, in users\u0027 notes files, they will use the\nID they define for each subsection.\n\nChange-Id: Ib7b4781de81e19d11a7acb863204067c0946827b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9f641d2efaa402bedacc0bf40d003041b1a6b2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9f641d2efaa402bedacc0bf40d003041b1a6b2b8"}]},"branch":"refs/heads/master"},"7071b671076c1b0b0de9fd99412fbfb72be1ff91":{"kind":"REWORK","_number":2,"created":"2023-02-10 20:32:23.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/65/873165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/65/873165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/65/873165/2"}}},"commit":{"parents":[{"commit":"919210c386a970ea0b28e4e7df3bb783c9f3895c","subject":"Refactor Section config to use NamedTuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/919210c386a970ea0b28e4e7df3bb783c9f3895c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-02-07 22:59:28.000000000","tz":-420},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-02-10 20:31:55.000000000","tz":-420},"subject":"Add support for subsections","message":"Add support for subsections\n\nSubsections will appear underneath their top-level section, in\nthe order defined in the config file. They will use a ^ or ~\nunderline rather than -.\n\nUsers define subsections by adding a number to the end\nof the section entry, either 2 or 3. It will be a subsection\nof the parent section above it. This syntax is simpler than others\nconsidered, like nesting lists inside a top-level section. It also\nmakes it easy for us to support deeper levels of nesting if we\u0027d like.\n\nLike normal sections, in users\u0027 notes files, they will use the\nID they define for each subsection.\n\nChange-Id: Ib7b4781de81e19d11a7acb863204067c0946827b\nStory: 2010375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/7071b671076c1b0b0de9fd99412fbfb72be1ff91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/7071b671076c1b0b0de9fd99412fbfb72be1ff91"}]},"branch":"refs/heads/master"},"ee73c9409e4bbceec565a42c7090a72455b1dcfc":{"kind":"REWORK","_number":3,"created":"2023-03-07 19:57:02.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/65/873165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/65/873165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/65/873165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/65/873165/3"}}},"commit":{"parents":[{"commit":"919210c386a970ea0b28e4e7df3bb783c9f3895c","subject":"Refactor Section config to use NamedTuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/919210c386a970ea0b28e4e7df3bb783c9f3895c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-02-07 22:59:28.000000000","tz":-420},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 19:56:39.000000000","tz":-360},"subject":"Add support for subsections","message":"Add support for subsections\n\nSubsections will appear underneath their top-level section, in\nthe order defined in the config file. They will use a ^ or ~\nunderline rather than -.\n\nUsers define subsections by adding a number to the end\nof the section entry, either 2 or 3. It will be a subsection\nof the parent section above it. This syntax is simpler than others\nconsidered, like nesting lists inside a top-level section. It also\nmakes it easy for us to support deeper levels of nesting if we\u0027d like.\n\nLike normal sections, in users\u0027 notes files, they will use the\nID they define for each subsection.\n\nChange-Id: Ib7b4781de81e19d11a7acb863204067c0946827b\nStory: 2010375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ee73c9409e4bbceec565a42c7090a72455b1dcfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ee73c9409e4bbceec565a42c7090a72455b1dcfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
