)]}'
{"id":"openstack%2Freno~876440","triplet_id":"openstack%2Freno~master~I6f290c3a49768252a12cc227050aa4e29ee243b4","project":"openstack/reno","branch":"master","topic":"fix-sort-order","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-03-07 15:45:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6f290c3a49768252a12cc227050aa4e29ee243b4","subject":"POC: Fix sort order for stable/2023.1 release","status":"ABANDONED","created":"2023-03-03 21:19:29.000000000","updated":"2023-03-07 15:45:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b1162479cacfa7c5c2c6a1139c351cbb0bc4ea7","_number":876440,"virtual_id_number":876440,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-03 21:43:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 21:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 08:02:28.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"}],"messages":[{"id":"a6441182e9d4ad45cb53f478916c89b807c7c941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-03 21:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a416da2748f784e52d8315ffa6d812dbebec192e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 21:43:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d6f094461f745dcbe171cdf64e06825\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50fdefeaf1cd458a86bba3bcdf97b3aa : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af098e76083b4c7b942a24e236ac947e : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12b6ac1a049549f6962d0c23cb81c7b5 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95fe51ea9bc94802817cc60152d794da : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ff41207e9143d9b9b94da5d148c825 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8d376ba6ab5da353caf185cc5d2e01086239a0fd","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-06 08:02:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b1162479cacfa7c5c2c6a1139c351cbb0bc4ea7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-07 15:45:22.000000000","message":"Abandoned\n\nA better version of this patch was merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4df19e8df087f496ea6f31867b52a046447769ae","revisions":{"4df19e8df087f496ea6f31867b52a046447769ae":{"kind":"REWORK","_number":1,"created":"2023-03-03 21:19:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/40/876440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/40/876440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/40/876440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/40/876440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/40/876440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/40/876440/1"}}},"commit":{"parents":[{"commit":"04233e0eae9d688a3eb0630443a30b860c04601e","subject":"Remove python-dev from bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/04233e0eae9d688a3eb0630443a30b860c04601e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-03 21:15:30.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-03 21:19:26.000000000","tz":60},"subject":"POC: Fix sort order for stable/2023.1 release","message":"POC: Fix sort order for stable/2023.1 release\n\nThe old behaviour is to use standard sort order for branch names, which\nplaces the recently created stable/2023.1 branch in front of all\nprevious stable/[a-z] branches. Fix this by adding a key function that\nwill make stable/2* sort as stable/zzz2*.\n\nTODO: make this more generic for people using branch_name_re other than\nstable/*\n\nChange-Id: I6f290c3a49768252a12cc227050aa4e29ee243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4df19e8df087f496ea6f31867b52a046447769ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/4df19e8df087f496ea6f31867b52a046447769ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
