)]}'
{"id":"openstack%2Freno~876784","triplet_id":"openstack%2Freno~master~I664156017ea24e819c6a7cffe6711376306b59fe","project":"openstack/reno","branch":"master","topic":"rm-print-for-deleted-notes","attention_set":{},"removed_from_attention_set":{"2472":{"account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"last_update":"2023-10-27 15:14:17.000000000","reason":"Change was submitted"},"35758":{"account":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"last_update":"2023-10-27 15:14:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I664156017ea24e819c6a7cffe6711376306b59fe","subject":"Remove noisy log statement for deleted release notes","status":"MERGED","created":"2023-03-07 22:14:59.000000000","updated":"2023-10-27 15:15:15.000000000","submitted":"2023-10-27 15:14:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"876784-rm-print-for-deleted-notes","meta_rev_id":"bbcda9782bef63321d54e16e3e2c5dd1b0df28cd","_number":876784,"virtual_id_number":876784,"owner":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 15:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-03-30 11:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-10-27 14:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-10-27 14:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 22:41:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 23:01:36.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2023-03-30 11:52:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-10-27 14:57:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1643ebb80527b69e7180abcd3cfaea03fab9d3b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 22:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef3a398ee30c5a67b917835476ab47c493f61f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 22:24:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e3e5be55e0c4e06af48857bce65877cabc9d308","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 22:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1925f2dc27d97af5237ebee869e378cf92049731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:41:15.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca66c041c6104f43bcaea708900f45a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13f0d6a913149a4a7c1d218605644a2 : FAILURE in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5f38d95e71f415dae4b50cb9d5c3e6d : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9545674165684c639ee70811e2f97391 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f06dbca3d48a4802861d6cec3f61a779 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6dbe6e756a4e069eb5412f60200967 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"18d2060c31f478a98dff6d803264f7ac3b574115","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2023-03-07 23:01:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dce5e600755152ee96215616891e802b78ade66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 23:26:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa094fd37e6adedb46b41644ac7fd2357667ae80","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-07 23:26:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fcf8853c6fa6c536174ef324c51402d0d8a07ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:33:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28928a24c2c34cb7bc18f9a673856018\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1827fc0d28e4f93a5149b089bff498a : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9718a79c930340028d3493730a7cf2f8 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50b5167dde2d40acbe4f8ccfce15a884 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f4676a13b94b3e9f235607c7aa0e28 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1557b22eda744b09e625ee81a979950 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"23c0258493bf52feb8236c627f5cc40b45e4d523","author":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"date":"2023-03-09 15:35:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c057ccf5cc496a42e9f3b632dcce561a5f6e781d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-03-30 11:52:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac41d00283ef4ca52dd88430323465c7b53e4999","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-10-27 14:57:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80a2c9963d69f6969c24392c52f45d79db21705a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 14:57:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09b5db8bf3f4bb7e1ceb1aaa48d0dd06e0f4a9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:14:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b6b46376324368bc75184b884edeea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d4ee2e97cb4806b4d123740467f662 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f210ca290484ab7958f96c12a0e1b61 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/629635171006422ab4c8142dfd18f631 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1069fb662bdd4b4ba6faad205da68178 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c98e8b19ff2d594e86e4353429cfda0b894236f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:14:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bbcda9782bef63321d54e16e3e2c5dd1b0df28cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:15:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f7863d0bb746e6a9e650667cbc911f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58cd83f3b8446f7b86830d9bbb190f9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"92a20ce518a504a8d312390785088574d200b897","revisions":{"57c1fcd8685e26bed4d5f79697691e9c04fb2141":{"kind":"REWORK","_number":1,"created":"2023-03-07 22:14:59.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/84/876784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/84/876784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/84/876784/1"}}},"commit":{"parents":[{"commit":"64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c","subject":"Order branches according to OpenStack release names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 22:10:29.000000000","tz":-360},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 22:12:59.000000000","tz":-360},"subject":"Remove noisy log statement for deleted release notes","message":"Remove noisy log statement for deleted release notes\n\nI believe this `print()` statement was not intentional, as it\u0027s the only `print` to `sys.stderr` in the entire project. We already are logging to debug level, so we don\u0027t need to also print.\n\nThis issue causes ~20 log lines for Qiskit Terra any time we run Reno.\n\nChange-Id: I664156017ea24e819c6a7cffe6711376306b59fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/57c1fcd8685e26bed4d5f79697691e9c04fb2141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/57c1fcd8685e26bed4d5f79697691e9c04fb2141"}]},"branch":"refs/heads/master"},"2aeb414d17c3fd5765f98fd0a35704719497d43e":{"kind":"REWORK","_number":2,"created":"2023-03-07 22:24:51.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/84/876784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/84/876784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/84/876784/2"}}},"commit":{"parents":[{"commit":"64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c","subject":"Order branches according to OpenStack release names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 22:10:29.000000000","tz":-360},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 22:24:42.000000000","tz":-360},"subject":"Remove noisy log statement for deleted release notes","message":"Remove noisy log statement for deleted release notes\n\nI believe this `print()` statement was not intentional, as it\u0027s the only `print` to `sys.stderr` in the entire project. We already are logging to debug level, so we don\u0027t need to also print.\n\nThis issue causes ~20 log lines for Qiskit Terra any time we run Reno.\n\nChange-Id: I664156017ea24e819c6a7cffe6711376306b59fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/2aeb414d17c3fd5765f98fd0a35704719497d43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/2aeb414d17c3fd5765f98fd0a35704719497d43e"}]},"branch":"refs/heads/master"},"92a20ce518a504a8d312390785088574d200b897":{"kind":"REWORK","_number":3,"created":"2023-03-07 23:26:18.000000000","uploader":{"_account_id":35758,"name":"Eric Arellano","display_name":"Eric Arellano","email":"ericarellano@me.com","username":"arellano"},"ref":"refs/changes/84/876784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/84/876784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/84/876784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/84/876784/3"}}},"commit":{"parents":[{"commit":"64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c","subject":"Order branches according to OpenStack release names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/64f649cd72c3ee34b0af17201e4db1d2e8e3cf8c"}]}],"author":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 22:10:29.000000000","tz":-360},"committer":{"name":"Eric Arellano","email":"ericarellano@me.com","date":"2023-03-07 23:12:07.000000000","tz":-360},"subject":"Remove noisy log statement for deleted release notes","message":"Remove noisy log statement for deleted release notes\n\nI believe this `print()` statement was not intentional, as it\u0027s the only `print` to `sys.stderr` in the entire project. We already are logging to debug level, so we don\u0027t need to also print.\n\nThis issue causes ~20 log lines for Qiskit Terra any time we run Reno.\n\nChange-Id: I664156017ea24e819c6a7cffe6711376306b59fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/92a20ce518a504a8d312390785088574d200b897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/92a20ce518a504a8d312390785088574d200b897"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
