)]}'
{"id":"openstack%2Freno~976737","triplet_id":"openstack%2Freno~master~I466883ddd2342f67248d60486b6447f4b4848b89","project":"openstack/reno","branch":"master","topic":"indexing","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-13 14:17:19.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I466883ddd2342f67248d60486b6447f4b4848b89","subject":"scanner: Decode sha before printing","status":"NEW","created":"2026-02-13 14:08:07.000000000","updated":"2026-05-20 18:55:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d49142012f1215d7192d41511fdcec21fbca00ba","_number":976737,"virtual_id_number":976737,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-20 18:55:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-13 14:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84320e8be3d4afd4d61cf5b72baefe5eeb687d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-13 14:08:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"804a6e5566fb093d2060d76b08bf0d3786d43a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 14:17:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8afd0d3b44a64246ba21a783f461bc64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5dce0139f6843dab5eacc0e54829266 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4af4a56d5f8447d18becf8257c87257d : FAILURE in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a631f8b9413426e834d4716d22e845f : FAILURE in 3m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e5ac93a660845f29a3fe6cfc2a9d115 : FAILURE in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8905416223641f5b3f90752ab34d0e5 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d45668a790a6eb351dc149ce1735796c69086703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 18:34:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d49142012f1215d7192d41511fdcec21fbca00ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 18:55:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3f7693d1c214f3c994e585cd94e87e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d627624f7bc4ceda71921debec45af8 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5bc41a0aae94862bbb4deb0ae8b7b44 : FAILURE in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0ae017f47fb442cb6b458ef3e74e1a0 : FAILURE in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/872a232cef3a41d6b6e101ecb1d7ac1f : FAILURE in 8m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/903e4174f5fb45ac870d5998bf7b90df : FAILURE in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b6230d0d214722a7277ea4003a920e : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec620606b2a94ed4b5ba67d18558981a69465245","revisions":{"855b84c139b150931c731fe709b43213f56f518f":{"kind":"REWORK","_number":1,"created":"2026-02-13 14:08:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/976737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/37/976737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/37/976737/1"}}},"commit":{"parents":[{"commit":"9d51b021d185521878d2685c9704d79301928625","subject":"Declare Python 3.12 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/9d51b021d185521878d2685c9704d79301928625"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 14:53:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 14:54:23.000000000","tz":60},"subject":"scanner: Decode sha before printing","message":"scanner: Decode sha before printing\n\nChange-Id: I466883ddd2342f67248d60486b6447f4b4848b89\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/855b84c139b150931c731fe709b43213f56f518f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/855b84c139b150931c731fe709b43213f56f518f"}]},"branch":"refs/heads/master"},"ec620606b2a94ed4b5ba67d18558981a69465245":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-20 18:34:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/37/976737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/reno","ref":"refs/changes/37/976737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/reno refs/changes/37/976737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/reno refs/changes/37/976737/2"}}},"commit":{"parents":[{"commit":"d4e513866a8b2ae4bae0a46d8491bc448e480e06","subject":"Migrate packaging configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/d4e513866a8b2ae4bae0a46d8491bc448e480e06"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 14:53:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-20 18:34:23.000000000","tz":0},"subject":"scanner: Decode sha before printing","message":"scanner: Decode sha before printing\n\nChange-Id: I466883ddd2342f67248d60486b6447f4b4848b89\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ec620606b2a94ed4b5ba67d18558981a69465245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/reno/commit/ec620606b2a94ed4b5ba67d18558981a69465245"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"855b84c139b150931c731fe709b43213f56f518f","theirs":"d4e513866a8b2ae4bae0a46d8491bc448e480e06","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
