)]}'
{"id":"openstack%2Frequirements~1000078","triplet_id":"openstack%2Frequirements~master~Ide5b9a452255a9f36ccb7057be833901271e89fe","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-30 08:09:11.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-30 09:50:46.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ide5b9a452255a9f36ccb7057be833901271e89fe","subject":"tox: Fail if ruff format makes changes","status":"NEW","created":"2026-08-07 10:07:20.000000000","updated":"2026-09-30 09:50:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56cab3c0af05637c9fe45229576e4d16050991f4","_number":1000078,"virtual_id_number":1000078,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-28 18:07:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-30 08:09:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-09-29 15:33:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-07 11:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-27 10:08:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-29 15:33:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8a2839c23742c8ba5c9afd7cdc90471bfb55d3f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-07 10:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec0b34a52ae6704824eca6f104993b514ae4912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:38:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88fa3fe378124b41b002591c93abfacf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef18e6c289884863a893ac4c42955b18 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7369922d9eb24c4f9329c1cec466db5f : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6096440a14c442178821d859b7b66a06 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bea789833d2487b93e81d3a34247c23 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78e1dca309674857ad98acce3d04882d : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9f39c51de5584339a70104b4ab82c50b : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b6485a0d53c4974a7e174393d83af52 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d233a5e995b402db518a6156321ff71 : SUCCESS in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b9abefa1beaf4e9e961fbe6ed3017318 : SUCCESS in 5m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ab2357c782b74fe9bb4668edea7eb514 : SUCCESS in 8m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b632cbdf3c834d44bc28022d320777ce : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f4437e2eb7643bb9ebfa1ddf514149c : FAILURE in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1295f37e1ec452ea9f518f5dadc0b39 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e97c8e1276d12789525c5ca4cc3c8e3c88850b97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-22 21:02:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c72b3246ea820564ee8992a6a2387b519ef205a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 22:35:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2edd1e2b6e67445a825278e8d228ed3e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/735fc49bbafc489ab395dd0dc6dde28a : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdfdf21408f4489f9137d46831f694c3 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743df59639244e22a9d8ea1caa1f2e6f : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b88eec58d481478289616da52bd8546c : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d15ea82f028a4432803bf033c0eb30f0 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a00cbb999d75488bb1a03d39261594c9 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8292e1b470e242379c579a6f1414a2a2 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3744e6bc04e2442294957dc06c4b9701 : SUCCESS in 3m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/53ce4fb20d6141dbb524c57976185cfb : SUCCESS in 9m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f9194f715d584bd78ddf2a1e7d2c870f : SUCCESS in 6m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/39284a76a7ae4ea6945067842c08f4d1 : SUCCESS in 7m 33s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/034c1eab8a224ef9b58b4b0888650c8b : SUCCESS in 8m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/151e1c44b43a4f9a8089858acaaa2f98 : SUCCESS in 58m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c50de43cfef49cb8acad43676d209e5 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0e9068e65d287d604f2bdecb20b104dbe5a3e168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 15:11:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9d8249486e975a7da4e9a5d23b569157b44d2e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7745eec43d804514a758e3421b3b3eb3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4ecd5d7f351646248f83e37514f332aa : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289028b7e78b49f49694317896e91a38 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22331241227f450ba7b219785186e36a : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6130af873c64664a6ed8fae8e83e7e5 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90dd0fee380b4a0b9699e3113b34b9fd : SUCCESS in 6m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b51335f1c5849b5b5abae4df52e1265 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ca36bf8a53d4e5a95faa0fcdb534732 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6eef61de6ba45be9d5d2bae8a94f23b : SUCCESS in 2m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a22e96858ea94fbcb4a352b9c4f20258 : SUCCESS in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a92434dc039f48e688145de085164946 : SUCCESS in 4m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f14336ed3a0f486f93a81de89dd16897 : SUCCESS in 4m 22s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/9ba0f56cc0194945a01d6c773da64180 : SUCCESS in 8m 21s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9be90afd486b416e96f642ce41f7b61b : SUCCESS in 31m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01227b3258174b5fb6e97ffea5720ec6 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6ef748a23615556dea71130188641056e3a93cba","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-27 10:08:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd7f715e5dc94fa6fbc7a0ad3b9b116eaa700b7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-28 15:22:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"697951282b0c7a8562e28f3f096acec0d7c8a2f9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-28 15:23:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"262b9de7058af2a26dde36dff34d5fb845312856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 18:07:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb4bdb913964a3e944d589e4e22adba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2abfafe514e44a0bf287741e30cc6d9 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2283aee177154794a80212042a237d43 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/223f26f3eef947a394f9779e5b8f2ec4 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8c8f70a84394802a09fee7c302d4430 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f4220332b1d459fb467881b552afc9e : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8130070eb4a447018dc92400f117f52e : SUCCESS in 3m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4756b4c7d2d14367994463cad796a554 : SUCCESS in 7m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9c1ed4be33474646b35c035c105d1e53 : SUCCESS in 6m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/da6a3ade30864bb3ba83b6f17adcfa47 : SUCCESS in 6m 17s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/b2837b37c07f48f3b3e75f1a5748c446 : SUCCESS in 9m 43s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3d3de6c43fa47b6bdd01d275fa41367 : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42ce52e4f67c4f16890d3ab490d5a0ec : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c013089e6e712b0afa4b0ed6693e59112cc81ad3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-29 15:33:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"479a5ca800f5524ec104cfb472b5a47ed9d20c3d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-30 08:09:11.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56cab3c0af05637c9fe45229576e4d16050991f4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-30 09:50:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a8e49ae926487bc9f80b5c80978a2a724130f2a2","revisions":{"1bb2a9f1dbbb10efd73df98e7c905524b920662b":{"kind":"REWORK","_number":1,"created":"2026-08-07 10:07:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1000078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/1000078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/1000078/1"}}},"commit":{"parents":[{"commit":"ddab97be8a4c5fe248db071fde9013ba530d7f72","subject":"Merge \"Add py-consul\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddab97be8a4c5fe248db071fde9013ba530d7f72"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-07 10:01:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-07 10:07:10.000000000","tz":60},"subject":"Enable ruff default rules","message":"Enable ruff default rules\n\nThe default rule set in ruff change in v0.20.0. Reflect the new one\nhere.\n\nWe also configure ruff to exit with non-zero code is anything changes,\nto ensure `tox -e pep8` fails in the gate if someone forgot to run this\nlocally.\n\nChange-Id: Ide5b9a452255a9f36ccb7057be833901271e89fe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bb2a9f1dbbb10efd73df98e7c905524b920662b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bb2a9f1dbbb10efd73df98e7c905524b920662b"}]},"branch":"refs/heads/master"},"6fc1e0291208031b2d8ac3005398d8b9c8cbce2d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-22 21:02:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1000078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/1000078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/1000078/2"}}},"commit":{"parents":[{"commit":"3353677727f8351bc58ba98efafe6bc884a0257c","subject":"Merge \"update constraint for os_vif to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3353677727f8351bc58ba98efafe6bc884a0257c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-07 10:01:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-22 21:02:45.000000000","tz":0},"subject":"Enable ruff default rules","message":"Enable ruff default rules\n\nThe default rule set in ruff change in v0.20.0. Reflect the new one\nhere.\n\nWe also configure ruff to exit with non-zero code is anything changes,\nto ensure `tox -e pep8` fails in the gate if someone forgot to run this\nlocally.\n\nChange-Id: Ide5b9a452255a9f36ccb7057be833901271e89fe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fc1e0291208031b2d8ac3005398d8b9c8cbce2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fc1e0291208031b2d8ac3005398d8b9c8cbce2d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ddab97be8a4c5fe248db071fde9013ba530d7f72","ours":"1bb2a9f1dbbb10efd73df98e7c905524b920662b","theirs":"3353677727f8351bc58ba98efafe6bc884a0257c","merge_strategy":"recursive","contains_conflicts":false}},"7c3a8e6545d5916bbc474fe735b95ec714a074d3":{"kind":"REWORK","_number":3,"created":"2026-09-18 15:11:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1000078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/1000078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/1000078/3"}}},"commit":{"parents":[{"commit":"3353677727f8351bc58ba98efafe6bc884a0257c","subject":"Merge \"update constraint for os_vif to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3353677727f8351bc58ba98efafe6bc884a0257c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-07 10:01:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 15:11:06.000000000","tz":60},"subject":"Enable ruff default rules","message":"Enable ruff default rules\n\nThe default rule set in ruff change in v0.16.0. Reflect the new one\nhere.\n\nWe also configure ruff to exit with non-zero code is anything changes,\nto ensure `tox -e pep8` fails in the gate if someone forgot to run this\nlocally.\n\nChange-Id: Ide5b9a452255a9f36ccb7057be833901271e89fe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c3a8e6545d5916bbc474fe735b95ec714a074d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c3a8e6545d5916bbc474fe735b95ec714a074d3"}]},"branch":"refs/heads/master"},"a8e49ae926487bc9f80b5c80978a2a724130f2a2":{"kind":"REWORK","_number":4,"created":"2026-09-28 15:22:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1000078/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/1000078/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/1000078/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/1000078/4"}}},"commit":{"parents":[{"commit":"e940761cf4f0d2505ce95f92ce12fb568d2de67a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e940761cf4f0d2505ce95f92ce12fb568d2de67a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-28 15:00:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-28 15:20:33.000000000","tz":60},"subject":"tox: Fail if ruff format makes changes","message":"tox: Fail if ruff format makes changes\n\nWe were not checking the result of the ruff commands, which meant it\ncould make changes but we wouldn\u0027t see failures in CI. Fix this by\nadding the missing arguments to ensure this happens and address issues\nthat have crept in in the interim.\n\nChange-Id: Ide5b9a452255a9f36ccb7057be833901271e89fe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8e49ae926487bc9f80b5c80978a2a724130f2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8e49ae926487bc9f80b5c80978a2a724130f2a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
