)]}'
{"id":"openstack%2Frequirements~1001895","triplet_id":"openstack%2Frequirements~master~I109b88b3cd9bfa0ec595712cf8a10c558586650e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I109b88b3cd9bfa0ec595712cf8a10c558586650e","subject":"update constraint for os-traits to new release 3.9.0","status":"MERGED","created":"2026-08-21 19:45:59.000000000","updated":"2026-08-22 06:05:29.000000000","submitted":"2026-08-22 06:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001895-new-release","meta_rev_id":"f8decae9fe9b9489073e018c298f7ebf05551480","_number":1001895,"virtual_id_number":1001895,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-22 06:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-08-22 05:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-08-22 05:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-21 20:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-22 05:01:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"636cd06918233bbeb513a561f44ccfb9423a717a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-21 19:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7acaea27a625785f30e88a41a12ba29a1b8ac3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 20:36:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54de53f8ef704ba9996e46498f2cd2bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0863209ba740d59c3e65c415657906 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4688dbcd0e794e89a2a5b98e9b8a527f : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/304cf41c6fc348c3b0b0a1c332413f48 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cb9af6f9e254579abe9726d6f89f904 : SUCCESS in 6m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98e4442796954751bf5c37c30742fa41 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37359a4b84c249ca955f28edb0c2145f : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5997bf1504f744d7b58d558746179504 : SUCCESS in 2m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/546349b14022491d8625a54e45dd1465 : SUCCESS in 3m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e8d826ab7f7c4775b12252b0adbf478b : SUCCESS in 5m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/efc4c84e3670443e8077adb849de9b95 : SUCCESS in 3m 09s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/e6f3e8e7e35e44678f0ede1b45c9c4a6 : SUCCESS in 5m 17s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/a45f7af054934ba59e8f429ac45c8679 : SUCCESS in 4m 46s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/685ee4b1ce1a4f39b001f3cc03482f50 : SUCCESS in 4m 46s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/e41256825982436eba69f91205c51f11 : SUCCESS in 3m 47s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f144e0e638054330880ef92e663068bc : SUCCESS in 7m 01s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/199e80a489c643339d645e464d08f7c8 : SUCCESS in 4m 49s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/20b1070fe7a34a0895101354982f2734 : SUCCESS in 5m 11s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/1e3fa3eecb1947fda209779bcaec4a12 : SUCCESS in 3m 56s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/97418525d2414b7d82cf59fe36b59a41 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54e7c18faa8140478c8a1801fadb759e : SUCCESS in 5m 58s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/83835275dd9c4ff88d491f5b0f1101cc : SUCCESS in 5m 20s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6c29ce79d9824758b584b96d91dcf893 : SUCCESS in 6m 50s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8822a3f4e0094bd989601048cad37b35 : SUCCESS in 2m 17s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/c5aaf8c6da3e4a1ba818f9b848a62a22 : SUCCESS in 6m 47s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c0f643573bdf44ab935c27a3fda59dfc : SUCCESS in 3m 43s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/2e1641284c4f4ff4929a8705216aaa4e : SUCCESS in 3m 15s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/04630b1788494f08af4ec381d7a12565 : SUCCESS in 11m 02s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/e906f89b00c14f16b11b0ed2e3364191 : SUCCESS in 20m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a666872955454a24b94c91af3cbc74a2 : SUCCESS in 6m 52s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/af557f32a63c403785db795dd07fa9df : SUCCESS in 13m 29s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/dbea4ae48c024e6fb40ea19d7f2f4c86 : SUCCESS in 3m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96da665263894926bad82be65b1b36a0 : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4aa27beeca01426381b5412b5b12c5ba : SUCCESS in 25m 20s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/7c29d2086e6d499aa23c99493ddf1a53 : SUCCESS in 4m 03s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/44acdfdf0f3d43658dd445b1c7f50e53 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b8949c0d5a346c58ab47728f0f268d2 : SUCCESS in 5m 17s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/12a088d3e34d4420a59a429520183e98 : SUCCESS in 12m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/51df4e638d8744efb95c8c75d0b7ae84 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1806ac1aad894431bd219a3ea4e5a6e6 : SUCCESS in 48m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4993ec03048495c949a4be55c8d7c32 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":1},{"id":"93d27215d7d53f909e212290c78c8accd8d5ff9d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-22 05:01:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a4ba98639546959829c1dcb76e09c907ab8738c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 05:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5ba20ea8baa02ecb0182324c276f97c964b7f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:04:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ffba625b394a31991afe0f75dc85c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3e041e923042baae9d504c8bca7d12 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63b4dd9bbec4922a263bddbb8c9b292 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/911b14ef1df24720b059307b19be131f : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/008c933b138940b1a6a07b25e26937f7 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/404ab1cddd2640d4acdb905e012f9c88 : SUCCESS in 2m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72a209ea0d3541dd824818051645d2e9 : SUCCESS in 4m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5bef819fd1e84b54b5203c0334d7fe68 : SUCCESS in 8m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/507829d64d3440f6b825437a8aa92824 : SUCCESS in 7m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8db5c6c67446418a9f45ab3e91c250b7 : SUCCESS in 7m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/d44c8ae34a884e29a2c8566d2375bbcd : SUCCESS in 9m 35s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/5126c6da865d47dba8cb795e1ab1cf0a : SUCCESS in 4m 28s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/42fedf683e9a4087846a5381e955666f : SUCCESS in 4m 26s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/baec9b363ed84e5fa8f6dd1959397d2f : SUCCESS in 5m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a863ad71511b4b198962c15c0cda5876 : SUCCESS in 15m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c291b887ce8a4cc787491ab2f646fa32 : SUCCESS in 6m 07s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/c5e43a75f6994868b140549dd631e2c3 : SUCCESS in 6m 57s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/41745561d52a4f9596beb5d561ab882f : SUCCESS in 12m 48s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/fe8470f2053e4cc28fa9ad9b2a11f888 : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8a8ef0117b544878b7fdc13191ce0d6e : SUCCESS in 8m 13s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/06a66236fc8e43beb275870c31dc4b37 : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eec2fb8b1f384ea8a4c39961efb54aff : SUCCESS in 5m 45s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0c3c33d72fd74449a1cf767fde46c6a4 : SUCCESS in 12m 32s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d2be718ed7a9448a92d4900ff9c7afb9 : SUCCESS in 2m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7470dc2e8b644dce98afd83cdb43f745 : SUCCESS in 2m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/167688ec95d145f6bfd5d1143d7efec9 : SUCCESS in 25m 58s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e7026aa5262848009d5d84b5cb4e50b4 : SUCCESS in 5m 00s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bd5b907647fe4623bda3ab63f1194393 : SUCCESS in 3m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c217f2f2b4ca45cca2bda90f0bc5187f : SUCCESS in 7m 01s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b36a5d5ab0d64730bacd0fbc6f2e7607 : SUCCESS in 8m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7e86c0dd5fb2444f84c2cc588396aea0 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9f6b2e2becd4388acb43a3b272af015 : SUCCESS in 58m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ceae569cb0542fc80dd34939ae82826 : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0920bec7eaee06b0fd09883b3e3a3a82b995429d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f8decae9fe9b9489073e018c298f7ebf05551480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:05:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a11e64d30a483e9d9793e95c1f3503\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bd7f62f5624cb6b9e99afff2820e4d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3519eae3119e8802e6bcf28b9ab102e30e2403f","revisions":{"d3519eae3119e8802e6bcf28b9ab102e30e2403f":{"kind":"REWORK","_number":1,"created":"2026-08-21 19:45:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/1001895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/1001895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/1001895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/1001895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/1001895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/1001895/1"}}},"commit":{"parents":[{"commit":"8150775c15f42a6aad869ca15fa6708f34a4feaf","subject":"Merge \"Bump alembic to latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8150775c15f42a6aad869ca15fa6708f34a4feaf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-21 19:35:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-21 19:35:57.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.9.0","message":"update constraint for os-traits to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Commit: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Change-Id: Ic2ff035372ccb6f2bc67d3b7245d0993f7927bb8\nmeta: release:Code-Review+1: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nChange-Id: I109b88b3cd9bfa0ec595712cf8a10c558586650e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3519eae3119e8802e6bcf28b9ab102e30e2403f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3519eae3119e8802e6bcf28b9ab102e30e2403f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
