)]}'
{"id":"openstack%2Frequirements~1001897","triplet_id":"openstack%2Frequirements~master~If46a652f89a6eeb9e32d40e54a9eb365ecbc667b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If46a652f89a6eeb9e32d40e54a9eb365ecbc667b","subject":"update constraint for os_vif to new release 5.2.0","status":"MERGED","created":"2026-08-21 20:02:28.000000000","updated":"2026-08-22 06:39:01.000000000","submitted":"2026-08-22 06:38:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001897-new-release","meta_rev_id":"f25e45b51988695b360a948726d48c95fe5495d2","_number":1001897,"virtual_id_number":1001897,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-22 06:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-08-22 05:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-08-22 05:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-21 21:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-22 05:01:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1f292fd5bbf3c64df716be2b5c21c8f846963f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-21 20:02:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a13da4d169b6db394794868f58c4ef3472f5afa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae220221e1b48f88073207bb9ac4a14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff21f2668764b528dd53dcf3bf5220e : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f22b0ad5f647f699bf892ab529b4e2 : SUCCESS in 1m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05abcee7408743eeb9fbfc1e74548aa5 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87b24e504e1e4960980a93f9146758ef : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/422c22c1c068483b8d6659935ae390b7 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d5b5ab681dc4924b6f6368de4d3c2f3 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42bbeaa6612a49aa8f9db0bafa416287 : SUCCESS in 2m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/421ec9da1e7841209aac90e352be0b43 : SUCCESS in 8m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/886fa52bbfce432384f30f1677dcc068 : SUCCESS in 5m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/33947614fa5e49c9a433cceced9e4b7d : SUCCESS in 5m 32s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/b8702f1e923c46e69d7ff9bafcbc7838 : SUCCESS in 7m 51s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/8bc7797f6ac84b6b92c39f4606fd9a6a : SUCCESS in 4m 38s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/cc8ec0daaebd437099b4ee6e27ed45f3 : SUCCESS in 4m 26s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/5d48205eb0c74b9daff262b6c94de350 : SUCCESS in 2m 52s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b3641eb709e548309dca80f31e76f6d6 : SUCCESS in 6m 45s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b929ac9b1cbc48f5861d6bff145a2f38 : SUCCESS in 6m 32s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/62e22cb4f9ad4a9cbcc4f9e37a9e2691 : SUCCESS in 4m 47s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9c0fc77ba591488596db4ec3d1e7991c : SUCCESS in 5m 32s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/3ab2646de85f4fa480b61fa5e6c0ae34 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e11c13fbc7f41dd9943b4ca299f62bf : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6eee832f8a4a4acf9f7ab152d89d72d9 : SUCCESS in 6m 04s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4873f6b3adcd47e5acd1a9fbbf51c5e9 : SUCCESS in 11m 38s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/d65a8b140fd44e37a8c38bc5659214ab : SUCCESS in 2m 16s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/72a7aa71093c40f5b12cd6902c740946 : SUCCESS in 8m 06s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/0013a79d6fca4cc5a0abca778c4c4fef : SUCCESS in 3m 45s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/310f4ddaf53c413aab6ae0c49ecbae0b : SUCCESS in 6m 07s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a0a5252e7a4943eaa2ad0d03ed6995c8 : SUCCESS in 9m 49s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/f253bb7ca7e94442935b296d3154b4af : SUCCESS in 16m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0bf88d54444b4a3c9af8e97050388cfe : SUCCESS in 3m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/003962a8a5144ed6bc10b901c5624b0f : SUCCESS in 8m 57s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/0c57d067810f4852869c3153a0102991 : SUCCESS in 4m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc092d10b26f439c9017c5d09cb2461b : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33d8b4a62f2c47a39f1047ee790ff280 : SUCCESS in 26m 44s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e287b9ab77464f139c16a02b33114570 : SUCCESS in 5m 13s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/f28764233a874861b22e7540ae77921c : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6aa9d2c440884af387c67f4eca2bdc2c : SUCCESS in 6m 21s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/bace675e3b58416c9902c427e638faec : SUCCESS in 18m 14s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e0f7136c86a5499ab58f851c04f7049c : SUCCESS in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8797a2b55717459c99ccaaf701789e5d : SUCCESS in 56m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d1c2e166db24de5b664bfd18497456f : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6d87107b2af20192dd09a755afc55f7b6d83550b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-22 05:01:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5355017340e663801079085828be03efbcdb31cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 05:02:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c8dfeaedadcd846d6bf6496eb93075d06514e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:38:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bfcd91c3e14b64bee49b68df6e3ec1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090c71ef4a5f4a43b7303fdc4c989357 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bafd9d16c7b54fdf824879a8ab346fa4 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18b49f93253e4b4e93f8c13abfef6b35 : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00bfe7a87d1c4899ae28531518d7cca9 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d5e569a90424e9084f8232d4422aa59 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d87adb1b7adc47ab99cf09dcd5f952af : SUCCESS in 2m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d066c0c726c48c9a6cb2350bb06f653 : SUCCESS in 6m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/67d2322256d245a8896b779ece313aa0 : SUCCESS in 9m 29s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0089ca302ce34703a397f266f8f077cf : SUCCESS in 5m 17s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/87367b4da02c4ca8b7ce3f113a4ab2d8 : SUCCESS in 12m 43s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/f089c9c180644508821913c4b1f8823a : SUCCESS in 4m 40s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/99a37fa8aeec405e88bee09970e641de : SUCCESS in 6m 33s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/dec978ad44864296857928a7b554ec10 : SUCCESS in 7m 15s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/43c312e5009548df88f58874e20153a2 : SUCCESS in 17m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bb990aebf0f44d5986df9ef5badfa9d : SUCCESS in 6m 48s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/6f0b0392760540f4bf09d89d76251730 : SUCCESS in 4m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/969028cafdf649e1b5df4e94926e67ab : SUCCESS in 11m 28s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/31009083a80343a282e0bac049a0d9dd : SUCCESS in 1m 45s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/bb6df2158f36446cb845332c59f76247 : SUCCESS in 8m 11s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0725c8e5483b4bc386c7b2fa5936d298 : SUCCESS in 21m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/69c698023bea45878c481b559ca05568 : SUCCESS in 6m 39s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/366d6e855a77402c81c7c896765c945d : SUCCESS in 10m 14s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1cd0c879f31b4bb7885eb9e79cd197d8 : SUCCESS in 2m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f1b89401ddc4eef9fd15d181208a057 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f807a001acf4813966f59515ab90e01 : SUCCESS in 26m 15s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d0d25ab792ab4e1fa6c2fe040105351e : SUCCESS in 4m 50s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5ca8306a7c6040c8bda389e88d17ddfc : SUCCESS in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b636ae6141042048203fc636913cbca : SUCCESS in 5m 13s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a086ea4635914bb18d790d69e93bb5b5 : SUCCESS in 8m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6c8f022b70c6424e9e1921a3234173ef : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46822237c06c4d8dad00a7c79ecebf0a : SUCCESS in 28m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dbad515c9884b2d94be1ab1afb292e0 : SUCCESS in 1h 30m 24s","accounts_in_message":[],"_revision_number":1},{"id":"53d043c5864edb144d7de00bd6cb3419fad6b8df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:38:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f25e45b51988695b360a948726d48c95fe5495d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:39:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d4f31e24a544a9b90d5fb2757b80d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5831d52ac81446ba2ca2d2e583a3f20 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af4e693b81ee1c5bd985ec3207cdc6f8004fccb6","revisions":{"af4e693b81ee1c5bd985ec3207cdc6f8004fccb6":{"kind":"REWORK","_number":1,"created":"2026-08-21 20:02:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/1001897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/1001897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/1001897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/1001897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/1001897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/1001897/1"}}},"commit":{"parents":[{"commit":"8150775c15f42a6aad869ca15fa6708f34a4feaf","subject":"Merge \"Bump alembic to latest version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8150775c15f42a6aad869ca15fa6708f34a4feaf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-21 19:52:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-21 19:52:24.000000000","tz":0},"subject":"update constraint for os_vif to new release 5.2.0","message":"update constraint for os_vif to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Commit: Sean Mooney \u003cwork@seanmooney.info\u003e\nmeta: release:Change-Id: I1e7a7662344eb6721eba68208595659b6774af5c\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: ribaudr \u003crene.ribaud@gmail.com\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: If46a652f89a6eeb9e32d40e54a9eb365ecbc667b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af4e693b81ee1c5bd985ec3207cdc6f8004fccb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af4e693b81ee1c5bd985ec3207cdc6f8004fccb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
