)]}'
{"id":"openstack%2Frequirements~1002116","triplet_id":"openstack%2Frequirements~master~I3b263c181207ce36a80d84ed961522e7e82bf8c4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3b263c181207ce36a80d84ed961522e7e82bf8c4","subject":"update constraint for oslo.db to new release 18.1.1","status":"MERGED","created":"2026-08-24 14:11:02.000000000","updated":"2026-08-26 09:09:31.000000000","submitted":"2026-08-26 09:08:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002116-new-release","meta_rev_id":"ba652c66d820c83f1bb5d38ab724c34c0d3aed1f","_number":1002116,"virtual_id_number":1002116,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 09:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-08-26 06:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-08-26 06:23:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 15:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-26 06:23:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"63c28a1975981b27fc0006c8c7973541d123c559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-24 14:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9155eeb19ed557a1bc0b16cb4f8adf95f8e9364d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3b03de03e9146f1a348e13d578e5431\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecca7b689fa4f6b864d4f8ae28afef4 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b6d4339a2d4bbf8a9bb5e36b0b083a : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/065ed8eda1c64f63a6362fbcb050de88 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95cbe9aa4fef4f34a8920de40d2ca5fb : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9e3dac66000436fb89ee0229bd5768d : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5fb3e1466e5649de8e6bfd03711654c0 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/787df28ad9d54145b616a1f38eb5816b : SUCCESS in 3m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b0d18a364bb145dcad1cc2c38c1fc692 : SUCCESS in 5m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1007c10ead8c49eda9b6abe73f2e9e0b : SUCCESS in 8m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fed7d71da49446cb86dd77e21be59208 : SUCCESS in 5m 35s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/928bd88d6d904400917fd29edc68c32d : SUCCESS in 9m 39s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e49a8d83b8654b39853a46361120df3b : SUCCESS in 4m 44s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/08a48e8c08e94a5083a8fd3e88e6a9d3 : SUCCESS in 4m 52s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/23befcaeebda46b7aef97e2fb6cc2599 : SUCCESS in 5m 15s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/eea4a772ca74413cad642931b1624b94 : SUCCESS in 7m 07s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/b4da142c26304e1ea58e6d8f05382d08 : SUCCESS in 5m 58s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/61d1c3cac26d46f692c96e31a8f6b08a : SUCCESS in 5m 12s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a7d831cdc3934da09d2b87d83fffc0d8 : SUCCESS in 6m 46s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/a4092e1946e94326af69decfd8c4b8d1 : SUCCESS in 18m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfe34d23721447f7a4482f75f5c30040 : SUCCESS in 8m 15s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/fab0d7754ded40eb988958b2276738ed : SUCCESS in 4m 14s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/abb805251229477ebfcb58dbcd6f584e : SUCCESS in 12m 24s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/109d81b01c084d7b923a80f0e44c747e : SUCCESS in 2m 30s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/bbbe7a95668f4b578319860be055d420 : SUCCESS in 4m 32s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d29c354b0059460a972042f931f6eddb : SUCCESS in 4m 42s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f42a158dd73344cb80956f8c1bd65efb : SUCCESS in 6m 47s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/cd1446480a494851ba53affdbc3847f4 : SUCCESS in 11m 20s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8a2f72f33f3f4d2993b46f4f62ca59c6 : SUCCESS in 17m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/538b7f4c72dd4261881e493e16caedf4 : SUCCESS in 9m 00s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/58f53e729480442789ffdb38461495cf : SUCCESS in 15m 31s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/7ce04e11310243ce9aa56cc0f8bcab26 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6863198d86945e88d4efec51a4fec71 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52f9a7a78e52478dbd0ce7e693cb67fd : SUCCESS in 26m 25s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/6049d22b9b4940e8baa9e0afc3a909e4 : SUCCESS in 5m 21s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/699ea9650c714db4834cba28c48fb1a5 : SUCCESS in 3m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b663b2c0eb74b8c88d3e2f4bbd8ee43 : SUCCESS in 7m 51s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/f19ec19e25dc4f4d9824b4949f1795cf : SUCCESS in 9m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ef237e326a546cfad3fcaa15a44abfd : SUCCESS in 6m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68db775db9ea4d21923bacb3ec708ce8 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ae4bc91bc024351af6c9f9a880c951a : SUCCESS in 1h 30m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8bad37a357738cd15db45e58b87cd1577fafd430","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-26 06:23:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4cc1c4009f60415bf5047b9e4b360783c9bfa8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:24:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6974876c3ee95caaa3042019623c07920896f762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:08:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4863bf8c7334e8ba02fc7621d016320\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df83dac0bb3d479a84c12fd45039a50b : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cf7a8ea25a42439e36be20e75dc1af : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81b0f23e6f6b4b8095be4add0f566174 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cec12347edb945b89c7095f80b185274 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/117acfe797b94315abf0cd0e97d1214f : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6011276b6c464c5e98f718522f5909f5 : SUCCESS in 3m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/76b817bf7b4c4d97bf509e6878a13cee : SUCCESS in 6m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a8447c9868374dd19eab4b23f376d488 : SUCCESS in 4m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fabe1b082d2c4287ae5f78ff0c711fc5 : SUCCESS in 5m 57s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/ea71153722b248678ef6102c98e62a99 : SUCCESS in 9m 02s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/7e932c2b495840fa96029c97627a382c : SUCCESS in 5m 39s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/60dfe56a6ec5463babfcca9d31b7c4fa : SUCCESS in 5m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/a4235e603b3e44d1941e7bb8829d3479 : SUCCESS in 6m 17s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/05807a7ba0b84ff2bade1d17ddf842b3 : SUCCESS in 18m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50e47e42729140d98ba5572390595865 : SUCCESS in 7m 04s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e4fda0d007c2478cb7515ab6de7cb890 : SUCCESS in 8m 06s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d4084b08b0e14e30941a822eda1f148e : SUCCESS in 11m 35s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/76f5d44c46f74ae09658599aac476091 : SUCCESS in 3m 23s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e9f3b50e2cab48e9be58afb4cc4a39e1 : SUCCESS in 8m 55s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/258e63707c8146159ed84ffa9b80ac11 : SUCCESS in 25m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f2bc5d00bd394c9c8f83c21890ea700f : SUCCESS in 6m 25s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/ab1176f91bb54de7949335a20e7bfacf : SUCCESS in 13m 54s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/dbceca6419ca474d955fc4f907867fd4 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d9cb3a1d0e6493b94bdcd45764df55f : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d9e458e87e94a3a9da65883c2916fa1 : SUCCESS in 29m 13s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/871a208e4f8f44789e17d411f6b2c1cc : SUCCESS in 5m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/50f4898609f140c9afb91218b4bfbfda : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af2550e9bc8a477c93bfb8115bb958a3 : SUCCESS in 8m 21s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/af90172af0a84230b355b09ae334fb90 : SUCCESS in 14m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79702b97a77346febef43a223b6d9a3e : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a29706ced24346afa2a178049ff9a494 : SUCCESS in 57m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3994b248cee34779b01246d3b4c65551 : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":1},{"id":"74bcf48aed12e1d95c0413aa90ffcdfe6a5f354e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:08:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba652c66d820c83f1bb5d38ab724c34c0d3aed1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:09:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b91fb764a444df7abe5d64881ca6b02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37a7425e1aa4e7fb7596c6f65c54d06 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b902baaa9bab2abcbd55a1029589807efaf57d59","revisions":{"b902baaa9bab2abcbd55a1029589807efaf57d59":{"kind":"REWORK","_number":1,"created":"2026-08-24 14:11:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/1002116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/1002116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/1002116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/1002116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/1002116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/1002116/1"}}},"commit":{"parents":[{"commit":"3353677727f8351bc58ba98efafe6bc884a0257c","subject":"Merge \"update constraint for os_vif to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3353677727f8351bc58ba98efafe6bc884a0257c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-24 14:01:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-24 14:01:00.000000000","tz":0},"subject":"update constraint for oslo.db to new release 18.1.1","message":"update constraint for oslo.db to new release 18.1.1\n\nmeta: version: 18.1.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Change-Id: Ia03db5af2af2fdd991ee64851334d88fa3385b5e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3b263c181207ce36a80d84ed961522e7e82bf8c4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b902baaa9bab2abcbd55a1029589807efaf57d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b902baaa9bab2abcbd55a1029589807efaf57d59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
